)]}'
{"id":"openstack%2Fneutron-lib~693412","triplet_id":"openstack%2Fneutron-lib~stable%2Fqueens~Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29","project":"openstack/neutron-lib","branch":"stable/queens","topic":"bug/1816485-stable/queens","hashtags":[],"change_id":"Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29","subject":"Add setproctitle support to the workers module","status":"ABANDONED","created":"2019-11-07 19:17:04.000000000","updated":"2020-04-03 17:30:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"35798397308854e2ad2ffb60b15b7620a699f970","_number":693412,"virtual_id_number":693412,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-07 20:38:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-11-07 19:17:04.000000000","_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 19:17:04.000000000","updated_by":{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},"reviewer":{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},"state":"REVIEWER"},{"updated":"2019-11-07 20:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e7b47a08b349fb94ee84bbd5110f826d3aa39e7","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-07 19:17:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cee4dc3dd87017ec57bc0632e94aa1326be6c38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 20:38:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60322a06e37548a4a136eb535ca584f4 : FAILURE in 1m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9204043aa02f4e45829f5d8e9f6dd420 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644d67f8189f4b82add9ec6b503d6e70 : FAILURE in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5794d027ada4a639973049fe1791f4d : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/df974c5ec4f2487ba7c4b63814b53e79 : SUCCESS in 3m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/313c5f036d474789bebfb8f38d85617c : SUCCESS in 1h 19m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2918262334c1499bb17cdd4b823d6518 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72f03b3728dc4cf7856bbc7ebb372a20 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdd145e162e1657ed8155b3206af4e9b767f14e9","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-04-03 17:30:03.000000000","message":"Abandoned\n\nCancelling backport b/c this does not meet backport rules.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"513ac1675972c934c0f5475c837a6df311988d4d","revisions":{"513ac1675972c934c0f5475c837a6df311988d4d":{"kind":"REWORK","_number":1,"created":"2019-11-07 19:17:04.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/12/693412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/693412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/693412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/693412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/693412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/693412/1"}}},"commit":{"parents":[{"commit":"37d25a3dd90083929828fded4c94b9248bfee8be","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37d25a3dd90083929828fded4c94b9248bfee8be"}]}],"author":{"name":"Doug Wiegley","email":"doug@parkside.io","date":"2019-02-15 03:24:02.000000000","tz":-420},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-11-07 19:17:03.000000000","tz":-300},"subject":"Add setproctitle support to the workers module","message":"Add setproctitle support to the workers module\n\n- Set the process name of child neutron-servers to be something more\n  readily identifiable than today.\n- Enable by default for all users of the workers module, neutron will\n  have a conf setting for its workers.\n\nMatching neutron change:\nhttps://review.openstack.org/637019\n\nPartial-Bug: #1816485\nDepends-On: https://review.openstack.org/637024\nChange-Id: Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29\n(cherry picked from commit 2c1e09bfd5a275c74213d200a7455bbf46893a1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/513ac1675972c934c0f5475c837a6df311988d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/513ac1675972c934c0f5475c837a6df311988d4d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
