)]}'
{"id":"openstack%2Fneutron-lib~694680","triplet_id":"openstack%2Fneutron-lib~stable%2Fstein~I3d13a747f31e39c11f42e2f543eae3823cfbe356","project":"openstack/neutron-lib","branch":"stable/stein","topic":"rehome-test-tools-stable/stein","hashtags":[],"change_id":"I3d13a747f31e39c11f42e2f543eae3823cfbe356","subject":"rehome used neutron.tests.tools","status":"ABANDONED","created":"2019-11-16 19:39:53.000000000","updated":"2019-11-18 19:00:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32921379291cdb03f92372572e01613630e816de","_number":694680,"virtual_id_number":694680,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-18 18:59:54.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-16 20:53:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2019-11-18 13:22:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-11-18 12:01:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2019-11-18 16:14:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"date":"2019-11-16 19:39:53.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-16 19:39:53.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-11-16 20:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-18 12:01:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-11-18 13:22:57.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2019-11-18 16:14:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"d490161b5db392ab90e60a410461b90d6873c5d0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-16 19:39:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"04eb2389dd0c38e701c3a2bc24739130a94ebb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 20:53:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d25f6cd91d430aadff74c282be4ffc : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44465d72ac2f46e49fdbe6c58071366d : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9d309ae9ed5469ba5a13cc92d60dfbc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf36d3df9dff43b597186ec428a81ca3 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58ce0144cf1148c9b2ecd91d5387f138 : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d296e31fcfa9493f928df9b8eb30ca5b : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06e321ed85847d4bbeb78066a91eb25 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84f2f99c365242c683e64df1cfb6e9df : SUCCESS in 3m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f08b98c507484d8d892b1336a038ff94 : SUCCESS in 1h 12m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32d5e7930ae74f0da0c600241f03bbe7 : SUCCESS in 1h 08m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c192cf1f1ffb40febac9192c84ca2177 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5f824901e6f7513df30a52b45084a5b03c6a456c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 11:55:16.000000000","message":"Patch Set 1: Code-Review-1\n\nIs this really necessary to be backported?","accounts_in_message":[],"_revision_number":1},{"id":"c30d104018375e94bfd7210fd51bdaa5bcdb9f0e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-18 11:57:09.000000000","message":"Patch Set 1:\n\n\u003e Is this really necessary to be backported?\n\nIt is for https://review.opendev.org/#/c/694317/","accounts_in_message":[],"_revision_number":1},{"id":"09465f4df4fa95166cde2b0674a58d608833d2a2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 12:01:54.000000000","message":"Patch Set 1: Code-Review+1\n\nok.","accounts_in_message":[],"_revision_number":1},{"id":"b4425e869ea8871748137bb4c2ff02145e147f5c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-11-18 13:22:57.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e \u003e Is this really necessary to be backported?\n \u003e \n \u003e It is for https://review.opendev.org/#/c/694317/\n\nIf we do backports like this in neutron-lib then we also need to release the stable branch with a new bugfix version and bump requirements in the corresponding neutron backport, don\u0027t we?\n\nIt seems simpler to me to add the few lines of code needed to the backported neutron change.","accounts_in_message":[],"_revision_number":1},{"id":"a440fb224fa186504f3c0e26581ad5021aa7541f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-18 16:14:41.000000000","message":"Patch Set 1: Code-Review-1\n\nI would agree with Bence, the neutron change should be taking care of this, we shouldn\u0027t introduce this change in a stable branch.","accounts_in_message":[],"_revision_number":1},{"id":"dc790f0f83ad9f0cee0eea67040e26327ca33aad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-18 18:59:54.000000000","message":"Patch Set 1:\n\n\u003e I would agree with Bence, the neutron change should be taking care\n \u003e of this, we shouldn\u0027t introduce this change in a stable branch.\n\nThanks Brian and Bence! I\u0027ll link those comments in the original patch.\n\nI\u0027ll abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"1d34be94acd67be1bbaa62e1a05cb7c9724dfad0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-18 19:00:03.000000000","message":"Abandoned\n\nAs commented in the bug.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4674fd8408a496e5fa8a5a192aa2f49cce5bbf4e","revisions":{"4674fd8408a496e5fa8a5a192aa2f49cce5bbf4e":{"kind":"REWORK","_number":1,"created":"2019-11-16 19:39:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/694680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/80/694680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/694680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/694680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/80/694680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/80/694680/1"}}},"commit":{"parents":[{"commit":"bba52345ee219a85f41bc87ed7ff20f5de2d9b1c","subject":"Filter placement API endpoint by type too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bba52345ee219a85f41bc87ed7ff20f5de2d9b1c"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-03-08 19:16:27.000000000","tz":-420},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-16 19:39:53.000000000","tz":0},"subject":"rehome used neutron.tests.tools","message":"rehome used neutron.tests.tools\n\nThis patch rehomes the OpenFixture class and reset_random_seed function\nboth which are from neutron.tests.tools as they are used by consumers\ntoday [1].\n\n[1] http://codesearch.openstack.org/?q\u003dfrom%20neutron.tests%20import%20tools\n\nChange-Id: I3d13a747f31e39c11f42e2f543eae3823cfbe356\n(cherry picked from commit ef9c2ae11198172b9980f033721bbcaf3b797909)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4674fd8408a496e5fa8a5a192aa2f49cce5bbf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4674fd8408a496e5fa8a5a192aa2f49cce5bbf4e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
