)]}'
{"id":"openstack%2Fneutron-lib~696308","triplet_id":"openstack%2Fneutron-lib~master~Ie0cccc5bd736b0302f956a7f01a9c16a8cc75200","project":"openstack/neutron-lib","branch":"master","topic":"amend-subnet-dns-ext","hashtags":[],"change_id":"Ie0cccc5bd736b0302f956a7f01a9c16a8cc75200","subject":"Change the requirements for the subnet_dns_publish_fixed_ip extension","status":"MERGED","created":"2019-11-27 13:46:09.000000000","updated":"2019-12-04 02:20:00.000000000","submitted":"2019-12-04 02:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696308-1575425940069-82587bb4","meta_rev_id":"8afc8c5f4ca449644d46e8653df410a470ab9ad0","_number":696308,"virtual_id_number":696308,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-27 19:26:54.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 02:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-12-04 00:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-11-28 21:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2019-12-04 00:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 21:16:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-04 00:33:21.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-12-04 02:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4f68f4151e8a4265bdf8c9cff5a87c00617e513","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-27 13:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1818d8ec40f8909800a3fb2dc1610195e3838499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 15:57:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ffbcb52d1041be886abe029468e0a1 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eb14a2b547b4d8c92d453d8a984d88b : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55f6f88d35e435eafe9d199d1cfa032 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea0587ecc9b4fc68bab00d108a8f891 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70f1ca5e85f945bda5635855de6ab709 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a488d1dfeb49c18b69f701cb22a2d8 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f0db1d5d03e42ab804f43e4941b2101 : SUCCESS in 4m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9fbc29503d7142fda8456e7b30dd3ab3 : SUCCESS in 1h 32m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3eee40d6ced449abbda923c46f8836f7 : FAILURE in 1h 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8a9ea7cb42db1f21022d17c512797027d174814e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-27 19:26:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddddfcd1cb23f4929e816118a1326d8cc7121f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:22:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8937c14b12d044b48a28a3bcccfb0416 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f76db87368c464b935f91c0dea29ed5 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2e9c19a1ebe47e1b7a53fc773e7b43d : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e08fa94d40141338ca9701960273acd : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82a56e959ba34708a72f71b3dd197a8b : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b146ce1f3c554a2a988597489adeec14 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4272506b0306484cb2386b19e0c8ae42 : SUCCESS in 4m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e56ac52356bf4272aa18b686f8325941 : SUCCESS in 1h 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d80722b233d4874869891152464f959 : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"125f1f60dd9704b94af8598ea1cc609f1bee8565","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-28 21:16:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa8ac37916ab5925326d10ad7838f768b193b967","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-04 00:33:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e192eedad0e2a727409045591a0973e0d3cc65cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 00:33:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3b49719a9291cbaf71d6ef2fd6261066e0a8a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:18:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2b833216b1485980e6794c204daa8c : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1acf957c94144191bdb61ff778e39c7f : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84416dff4a3640d1844a3fdcbe68b154 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47a9695df7934ba6888ac2ef4483105e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa60b745ad654bd98cac47c427f070b2 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8fb067e154a46adbfcb33b415669bed : SUCCESS in 4m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdb6eb4b8c094f049815b7823078d15a : SUCCESS in 1h 41m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a7aa344117a4b4e954799ebe56459b1 : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0e6750f240befc80e601c46666956ca657fd5470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:19:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5fde0472d6d8f6a48a7313e3b7aef9c8690f1fa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:20:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37f4c33ee10421999f74f09055440eb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c352a65f4639d34b73c8500197a996ca4c582b4","revisions":{"7c352a65f4639d34b73c8500197a996ca4c582b4":{"kind":"REWORK","_number":1,"created":"2019-11-27 13:46:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/08/696308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/696308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/696308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/696308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/696308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/696308/1"}}},"commit":{"parents":[{"commit":"85205e9f08dddee9246969fbc66a6ae366ae1946","subject":"Merge \"Remove \u0027interconnection\u0027 API extension definition (neutron-interconnection)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/85205e9f08dddee9246969fbc66a6ae366ae1946"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-11-27 13:28:49.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-11-27 13:46:07.000000000","tz":0},"subject":"Change the requirements for the subnet_dns_publish_fixed_ip extension","message":"Change the requirements for the subnet_dns_publish_fixed_ip extension\n\nDuring the implementation of the subnet_dns_publish_fixed_ip\nextension [0] it was noticed that it will be better to implement just\na single stack of dns_integration extensions instead of having the\noption of combining them flexibly. So we need to have the new extension\nrequire the top of the current stack instead of just the basic extension.\n\n[0] https://review.opendev.org/#/c/662409/\n\nChange-Id: Ie0cccc5bd736b0302f956a7f01a9c16a8cc75200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7c352a65f4639d34b73c8500197a996ca4c582b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7c352a65f4639d34b73c8500197a996ca4c582b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
