)]}'
{"id":"openstack%2Fneutron-lib~697065","triplet_id":"openstack%2Fneutron-lib~master~I3264fe81fbc1ffc5d3af151e8e56ccc5345f8c9b","project":"openstack/neutron-lib","branch":"master","topic":"bp/s","hashtags":[],"change_id":"I3264fe81fbc1ffc5d3af151e8e56ccc5345f8c9b","subject":"WIP: Add agent_evacuate definitions","status":"ABANDONED","created":"2019-12-03 11:43:55.000000000","updated":"2020-05-19 15:26:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7161a0b4a8af078e86226ede14d236d71b329744","_number":697065,"virtual_id_number":697065,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-03 11:57:24.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-03 13:39:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 11:57:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-03 13:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4536af3a1279dd9ba6773f2da532967851d3004","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-12-03 11:43:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab400b83704e3900f4c171a0a2a5b34fdbd3cf24","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-03 11:57:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d5451c5eadc441e71b44e95eac79b0797ff1311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 13:39:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c09fb05bf5144de926bc8fa2cb6c2ab : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0963e817680b41429539b0445e3b1b97 : FAILURE in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a05aa74a8a594586a722e1d14980c6fd : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f3bfab0f5a4115ae5183038ef36898 : FAILURE in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86b499bbaaa2478aad1afd59db66a0f1 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bbb7954eba044828b73e74041a2c9dc : FAILURE in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e0f399558d34927b8fbadc5668122a4 : FAILURE in 4m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/652ff2e5b8f34bd383ceae07d3fa4142 : FAILURE in 1h 53m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb01871df48d4e6aae70f0705e6a734c : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d6530143c4e111cfbe05e649f0ffa66e33e5f0c6","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-19 15:26:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edcedb55299cff49940def2ae43b20b277ab5d4b","revisions":{"edcedb55299cff49940def2ae43b20b277ab5d4b":{"kind":"REWORK","_number":1,"created":"2019-12-03 11:43:55.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/697065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/697065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/697065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/697065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/697065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/697065/1"}}},"commit":{"parents":[{"commit":"85205e9f08dddee9246969fbc66a6ae366ae1946","subject":"Merge \"Remove \u0027interconnection\u0027 API extension definition (neutron-interconnection)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/85205e9f08dddee9246969fbc66a6ae366ae1946"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-12-03 11:42:59.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-12-03 11:43:52.000000000","tz":60},"subject":"WIP: Add agent_evacuate definitions","message":"WIP: Add agent_evacuate definitions\n\nDo not merge, just a first try...\n\nSee:\n\nhttps://blueprints.launchpad.net/neutron/+spec/evacuate-agent-resources\n\nChange-Id: I3264fe81fbc1ffc5d3af151e8e56ccc5345f8c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/edcedb55299cff49940def2ae43b20b277ab5d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/edcedb55299cff49940def2ae43b20b277ab5d4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
