)]}'
{"id":"openstack%2Fneutron-lib~698741","triplet_id":"openstack%2Fneutron-lib~master~Ibb78270c54d1775b0596a00ae652f6f0b7f7dff8","project":"openstack/neutron-lib","branch":"master","topic":"dontdelete/rehome_security_groups","hashtags":[],"change_id":"Ibb78270c54d1775b0596a00ae652f6f0b7f7dff8","subject":"WIP: rehome security_group api-def into neutron-lib","status":"ABANDONED","created":"2019-12-12 15:26:32.000000000","updated":"2020-07-09 06:28:22.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"49d9b559bf3ac9d140686b6bf5828f8a6e32c842","_number":698741,"virtual_id_number":698741,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-13 15:13:53.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-12 15:55:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 15:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a65b140bb3a15f18d69f32ea75f507edc386819b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-12 15:26:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8f85d03c451fe68b1a74f43869197f18d993db8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-12 15:29:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3931453317d605da7ce2e4fc9ab705acb03cd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:55:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece1fe61c7d74397bb7982b45395ad4b : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e9555fee974ff68331e200b8c112d2 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/227bc4171fc14e65853d02735f5125b1 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95efe589714c4b699f8702fcecc82e03 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a9986c3f8274f73b4f9ef8563fb695f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a8c687fd5f345c5b94e496784d6908c : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/360dbdbcf2514c6fa0d36ef20e797ec0 : SUCCESS in 6m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c696d56656ac465f96deb81fc512593d : FAILURE in 25m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4363a03828374a9abc9e18a7754bc422 : FAILURE in 24m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3496ee23b5524616be851aa4c0629bd5 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d4a4854c951a5cb9efa511a7d7ac4632d6da6ea8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-13 15:13:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"96f79623f438f4477497ced9035544bea2010539","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-09 06:28:22.000000000","message":"Abandoned\n\nBased on this policy change: https://review.opendev.org/739147 the rehoming lost priority. Later it should be reopened","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa6dc7c5455e936d5c4a9e01f03f521a5149bfd7","revisions":{"aa6dc7c5455e936d5c4a9e01f03f521a5149bfd7":{"kind":"REWORK","_number":1,"created":"2019-12-12 15:26:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/41/698741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/698741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/698741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/698741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/698741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/698741/1"}}},"commit":{"parents":[{"commit":"3b5413fbc453d8ac0947382596735bcdc8fc56a6","subject":"Merge \"Add description field in port forwarding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b5413fbc453d8ac0947382596735bcdc8fc56a6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-12-12 15:03:08.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-12-12 15:25:57.000000000","tz":60},"subject":"WIP: rehome security_group api-def into neutron-lib","message":"WIP: rehome security_group api-def into neutron-lib\n\nThis patch rehomes neutron.extensions.securitygroups API definition and\nassociated exceptions into neutron-lib.\n\nUnit tests and release note are also included.\n\nChange-Id: Ibb78270c54d1775b0596a00ae652f6f0b7f7dff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aa6dc7c5455e936d5c4a9e01f03f521a5149bfd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aa6dc7c5455e936d5c4a9e01f03f521a5149bfd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
