)]}'
{"id":"openstack%2Fneutron-lib~700754","triplet_id":"openstack%2Fneutron-lib~master~I69904a235121876a14dbb3e4cf1beac614dcd5e9","project":"openstack/neutron-lib","branch":"master","topic":"tag-ports-during-bulk-creation","hashtags":[],"change_id":"I69904a235121876a14dbb3e4cf1beac614dcd5e9","subject":"Extension for tagging ports during bulk creation","status":"MERGED","created":"2019-12-30 00:49:00.000000000","updated":"2020-01-08 11:32:53.000000000","submitted":"2020-01-08 11:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700754-1578483013220-17819b5d","meta_rev_id":"a2efeccfd055f2d1a0c6c9a4680a1cdaf7cc9a46","_number":700754,"virtual_id_number":700754,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-04 00:19:49.000000000","_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 11:30:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-01-03 15:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-01-06 10:03:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-01-07 23:00:36.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-08 09:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 10:03:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-07 23:00:36.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-01-08 09:46:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-08 11:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45037ed455fb2a85c3c941218ed39aed4e8e194a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-30 00:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16b2b5100c4e45268d55fba0923f2959e7038a3e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-30 01:31:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c2911b52655a7f927a1d6cc0d99f3d09590016b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 02:56:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2fb83262f043dd8bffc3e927c4bc33 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de45f540359455e9e054c4cf8ceb100 : FAILURE in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/644d8ed3c6654185a850f6fc2baa7b91 : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92691b3f780342988db68ad872dda181 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a89f56652ae242f68a6bf48a2575d6f2 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb0fd2cc06ea401394cecef614734493 : FAILURE in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af3ecba20002437bac6b38f90ed11c77 : FAILURE in 3m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/29fc3c04e82445eb87c3f943a6a366d4 : SUCCESS in 1h 24m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f83fdf69ee314d87b7409f8481d843fc : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":2},{"id":"69842230e68d2348b284bbeb6cb1bc7344f03321","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-30 22:50:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"240910162ac94e018b4fce928ed9bc68af039ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 00:45:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3053b51825c45008a6a2587a6013097 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eabf6918458a414eb634740487cfc8f6 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5849026e5aba4c68b815bd41c930f66a : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b643559054c64b018dc7d251020bf800 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7afcec898c024bf498b65be30c9d16d8 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/501a4fcc37564abaa5cb80e5fa5af441 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b8f8cabf1cf4cbb8f5af17563a2a0d6 : SUCCESS in 5m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2785bf3164db4424821c35b18264ca3d : SUCCESS in 1h 24m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/810a9918e9804c7f9da782b6b653a9e6 : SUCCESS in 1h 21m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12002b86d7934ab08f230ab23ccb5a36 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"db8244563ad0ccd71b03c99b7a5a9ff84e0c83f4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-03 15:20:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d39ec358ef1bfec01b274e7f579fed8be4c1909f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-03 15:21:43.000000000","message":"Patch Set 3:\n\nWill You update api-ref https://docs.openstack.org/api-ref/network/v2/?expanded\u003dcreate-port-detail,bulk-create-ports-detail#ports in follow-up patch?","accounts_in_message":[],"_revision_number":3},{"id":"14f688f937bafb53ac860c69b93378684b5de4cb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-04 00:19:49.000000000","message":"Patch Set 3:\n\nYes!","accounts_in_message":[],"_revision_number":3},{"id":"382680c75a3873576b30437913e0134048115666","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-06 10:03:01.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e7514a69674fa8e08b967b8a7856e23cb638335","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-07 23:00:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"836c3d8979ee4282b110ce04bd0452063330dd01","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-08 09:46:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"372e331bd5531530c84df75b112d13d4b8ba9fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 09:47:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"52feb7f23c6b951b86c38097d5fd44ac24ae836d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:30:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cab8774e93da5f31c42a4f5e8e14f0505354f6e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:30:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ced3db2dcd40fbb007f9f25a7dc9e7 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fca61796359b40cf9b235e0ba87afdd6 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b0fa0db975c41218e5a791ca0fc434f : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de41a3e096ee4291b4b2fa443d76a084 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f90b5141bda146c08b1226c94135d36d : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53f0fdee1f944c31abe6e7fe0c218942 : SUCCESS in 6m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2bd89f0108014bf4a6bd3692903f91ad : SUCCESS in 1h 30m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64416249ea4c437d8b5a3bce7c840da7 : SUCCESS in 1h 40m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80b23ceef30646038998d86ef0434664 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"029e656bb7659e7f8a4dfdfb6fdd9c23856f71ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:32:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a47e77b4da6440a5ab08be77194e97cc : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/048e6b01cb484d2287ef33bc99918b14 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8673f033196771ef907314f823ca23ef9360134d","revisions":{"1b2fa4734c0431bcfe0a132bef0a30214813aee2":{"kind":"REWORK","_number":1,"created":"2019-12-30 00:49:00.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/54/700754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/54/700754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/1"}}},"commit":{"parents":[{"commit":"79cd582084c4b23626a83a2e9059cb5c6e6a29ee","subject":"Pass on missing implementation while filtering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/79cd582084c4b23626a83a2e9059cb5c6e6a29ee"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 00:48:48.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 00:49:00.000000000","tz":-360},"subject":"Extension for tagging ports during bulk creation","message":"Extension for tagging ports during bulk creation\n\nDefine a shim API extension to support the tagging of ports during bulk\ncreation.\n\nChange-Id: I69904a235121876a14dbb3e4cf1beac614dcd5e9\nRelated-Bug: #1815933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1b2fa4734c0431bcfe0a132bef0a30214813aee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1b2fa4734c0431bcfe0a132bef0a30214813aee2"}]},"branch":"refs/heads/master"},"dacfae183a129bf3121bed1cc951f868f6ecc69f":{"kind":"REWORK","_number":2,"created":"2019-12-30 01:31:27.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/54/700754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/54/700754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/2"}}},"commit":{"parents":[{"commit":"79cd582084c4b23626a83a2e9059cb5c6e6a29ee","subject":"Pass on missing implementation while filtering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/79cd582084c4b23626a83a2e9059cb5c6e6a29ee"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 00:48:48.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 01:31:19.000000000","tz":-360},"subject":"Extension for tagging ports during bulk creation","message":"Extension for tagging ports during bulk creation\n\nDefine a shim API extension to support the tagging of ports during bulk\ncreation.\n\nChange-Id: I69904a235121876a14dbb3e4cf1beac614dcd5e9\nRelated-Bug: #1815933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dacfae183a129bf3121bed1cc951f868f6ecc69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dacfae183a129bf3121bed1cc951f868f6ecc69f"}]},"branch":"refs/heads/master"},"8673f033196771ef907314f823ca23ef9360134d":{"kind":"REWORK","_number":3,"created":"2019-12-30 22:50:05.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/54/700754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/54/700754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/54/700754/3"}}},"commit":{"parents":[{"commit":"79cd582084c4b23626a83a2e9059cb5c6e6a29ee","subject":"Pass on missing implementation while filtering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/79cd582084c4b23626a83a2e9059cb5c6e6a29ee"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 00:48:48.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-12-30 22:49:52.000000000","tz":-360},"subject":"Extension for tagging ports during bulk creation","message":"Extension for tagging ports during bulk creation\n\nDefine a shim API extension to support the tagging of ports during bulk\ncreation.\n\nChange-Id: I69904a235121876a14dbb3e4cf1beac614dcd5e9\nRelated-Bug: #1815933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8673f033196771ef907314f823ca23ef9360134d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8673f033196771ef907314f823ca23ef9360134d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
