)]}'
{"id":"openstack%2Fneutron-lib~703241","triplet_id":"openstack%2Fneutron-lib~master~Ic6da328a6219c8a00d38602f6b539a2ff5be1b2e","project":"openstack/neutron-lib","branch":"master","topic":"remove_six","hashtags":[],"change_id":"Ic6da328a6219c8a00d38602f6b539a2ff5be1b2e","subject":"Remove library \"six\"","status":"MERGED","created":"2020-01-18 13:27:37.000000000","updated":"2020-01-22 23:36:53.000000000","submitted":"2020-01-22 23:35:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703241-1579736103645-46a14a57","meta_rev_id":"ba23ce3044aa07deff094464b5df2ee8351b29cb","_number":703241,"virtual_id_number":703241,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 23:35:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-01-22 22:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-01-22 14:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-22 22:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-18 13:32:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-18 13:32:55.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-01-20 11:35:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-21 05:44:53.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-22 14:56:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-22 22:09:37.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-22 23:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24d0defc1a42ddffe02f2cf8b486c8098ae70172","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-18 13:27:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"709ae91b7700f49f6aabf1a38b69435d6bd107a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/25fe23617bdc425bb72799521de20e81 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c116b58d10b941e4b2022e2c36ec83ef : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c31dc34563f4492c834c386acf05e358 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a92923492a9499dbd495abd6c3e633d : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb56bf3f0244cdeba6438a0ebaaaa44 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bfd06b5f6de45a5a9b4748e7bf592d0 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efcfc9c425434c5f9862f17b70b4288d : SUCCESS in 3m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d825795a43234e149f844bc0a9393bce : SUCCESS in 1h 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"91300696a956d35862408776c0bf61106b37ed2f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-18 17:15:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e24324c05fbb53c0ae1fdcf63f3ded344216f554","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-18 19:39:44.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"387d9348c83ffe7f7bf36bd57930ce93c22f5255","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-18 19:40:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f04749c87c7e098394655ed5b9459e1055db41d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 21:07:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/78ee20163824477487a60807f1ee9027 : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67517becee5c4566a301e13064ba0c61 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83c6e00112254f20baef758aade05de4 : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b68c1f04ab1493782d7397998e86808 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f69810bfe194f01aa20c541ba593916 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84794d0ef4194ac78b7d6905964b5e07 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1a6372270804bdaa9dd318dbf678748 : SUCCESS in 3m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9a619ffe77b4866a6cf7377ebda22b4 : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":2},{"id":"154e384ab01e93c68792e6c3d52e12af0a8aceaa","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-20 00:51:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ce2884d5e75bcb2a160de9bc0b965834fb35e301","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-20 07:34:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b4942c945a1e205987f9abc31b77281e460c875","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-20 08:37:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4e45942b0ee68f977c3ce5ca872278ebb8b2dc3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-20 10:51:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb1feba8b21a9a3fcaa3b2ea22a4b23e38ba9a05","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-20 11:35:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a107725937efcccc1b57ef439e20750b73705352","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-20 11:35:59.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"1cffcfda2fc3a6f21784ce2e19710b91734f02ae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-20 18:01:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f32138acad680e93f2ab4721bcca1bdb22c273f6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-20 20:31:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f60aefc2127133572be2f218d92002399d941fd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 05:44:53.000000000","message":"Patch Set 2: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2013ccedb232c6947520a8081597bd4074bed984","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-21 08:59:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35b4099d115ebc269a239b4c6fe34e2a46e738ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-22 13:15:09.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"76377607aefa9037b531a441482c77f98f8819bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-22 13:17:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c1716d84170057650750303711df33d57ba39d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 14:37:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/530aadb45eeb44579ac0b57c45b2fb42 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6c16d520844c27af525c6ef97fe32d : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76fefef814b94758af06fda741bfd110 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8998a83550c54a089ca5bd43af0fc321 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914316ac88ba45fba92f122559327fb3 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7960ad594766465d97a6909b5f20b010 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79bc1ebe16cd4521b104ac94fadfe0c3 : SUCCESS in 4m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a16f157a8074319a1b06df8d712937d : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b16c2b957e426c92404003a60748a2d5010a8d52","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 14:56:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6e132b0980f04b77e70044696f47e642bd98c496","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-22 22:09:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4ba8b4b86d84df770f513767c2dd00175812edc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:09:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"be4c8abc56bcf075dac4acf7b53ce60ab8fcf980","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 23:35:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d06dbec0e26c59efabc772c776b1d85fea9cbb75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 23:35:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c91d7e29aed34b16ad9069ff0d3a815a : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bce536752f148019dd835c83592c315 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5237c4ecd4b54ea7a9e931fd34c5cddb : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb76565d791d44ecaa3cd4087f5e167e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e64ef8a2a55f4b35bfec9b7834011bf1 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0940b3a851eb4f76b2399c791f21cc6e : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7d69ee08e4c4aaba40abad5dad792c3 : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":3},{"id":"309676f965b0d167c328488409ef365288d93422","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 23:36:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458ff1b7dc61402d9af212f0c4a6abca : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"478502b3dfb8c3ff66318169377251826563a398","revisions":{"1074e67ea55980b8a6265b714f2fb518d0df7bae":{"kind":"REWORK","_number":1,"created":"2020-01-18 13:27:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/703241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/703241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/1"}}},"commit":{"parents":[{"commit":"1d9d6d7bf5095efe96191741a93f37446ca4d1a4","subject":"Merge \"Fix new parameter added in \"qos_port_network_policy\" API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d9d6d7bf5095efe96191741a93f37446ca4d1a4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-18 13:23:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-18 13:27:26.000000000","tz":0},"subject":"Remove library \"six\"","message":"Remove library \"six\"\n\nNow Python2 is not supported, this compatibility library for both\nversions is not needed anymore.\n\nChange-Id: Ic6da328a6219c8a00d38602f6b539a2ff5be1b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1074e67ea55980b8a6265b714f2fb518d0df7bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1074e67ea55980b8a6265b714f2fb518d0df7bae"}]},"branch":"refs/heads/master"},"865b26d6ba9e42ac9f81c2ac2cade62919dec0a2":{"kind":"REWORK","_number":2,"created":"2020-01-18 19:40:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/703241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/703241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/2"}}},"commit":{"parents":[{"commit":"1d9d6d7bf5095efe96191741a93f37446ca4d1a4","subject":"Merge \"Fix new parameter added in \"qos_port_network_policy\" API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d9d6d7bf5095efe96191741a93f37446ca4d1a4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-18 13:23:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-18 19:38:17.000000000","tz":0},"subject":"Remove library \"six\"","message":"Remove library \"six\"\n\nNow Python2 is not supported, this compatibility library for both\nversions is not needed anymore.\n\nChange-Id: Ic6da328a6219c8a00d38602f6b539a2ff5be1b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/865b26d6ba9e42ac9f81c2ac2cade62919dec0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/865b26d6ba9e42ac9f81c2ac2cade62919dec0a2"}]},"branch":"refs/heads/master"},"478502b3dfb8c3ff66318169377251826563a398":{"kind":"REWORK","_number":3,"created":"2020-01-22 13:17:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/703241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/703241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/703241/3"}}},"commit":{"parents":[{"commit":"1d9d6d7bf5095efe96191741a93f37446ca4d1a4","subject":"Merge \"Fix new parameter added in \"qos_port_network_policy\" API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1d9d6d7bf5095efe96191741a93f37446ca4d1a4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-18 13:23:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-22 13:16:45.000000000","tz":0},"subject":"Remove library \"six\"","message":"Remove library \"six\"\n\nNow Python2 is not supported, this compatibility library for both\nversions is not needed anymore.\n\nChange-Id: Ic6da328a6219c8a00d38602f6b539a2ff5be1b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/478502b3dfb8c3ff66318169377251826563a398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/478502b3dfb8c3ff66318169377251826563a398"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
