)]}'
{"id":"openstack%2Fneutron-lib~706820","triplet_id":"openstack%2Fneutron-lib~stable%2Fstein~Id6e6d633b00237d8909160e7ed6f5e495399a252","project":"openstack/neutron-lib","branch":"stable/stein","topic":"bug/1828543","hashtags":[],"change_id":"Id6e6d633b00237d8909160e7ed6f5e495399a252","subject":"placement client: fix routed prov networks working","status":"ABANDONED","created":"2020-02-10 13:00:45.000000000","updated":"2020-08-04 13:58:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c6e701748be6460a6893cf76ce5fa0cce3d296f","_number":706820,"virtual_id_number":706820,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-10 14:43:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"all":[{"value":2,"date":"2020-02-11 20:36:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-02-11 01:39:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":-1,"date":"2020-02-12 11:37:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 14:43:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 01:39:31.000000000","updated_by":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"reviewer":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"state":"REVIEWER"},{"updated":"2020-02-11 20:36:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"878f1ef192ba47481a7d7d712d08a5b9114c0446","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-10 13:00:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5e56e6bd8b07c47fe69e79b923c31cdc94ef5b61","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-10 13:01:22.000000000","message":"Topic changed from bug/1828543-stable/stein to bug/1828543","accounts_in_message":[],"_revision_number":1},{"id":"ed0d2404cbeaccb8bdbae70eee9bb70058baff50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:43:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed71ca9bc2a42cca1f5de6b91d61d62 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb98d9671cba422daf9bfc3c1087a5d5 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4816e77b188842b4b7d6ed39b66dca56 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e0c8363c004bb08cb0fcfee0bc5a0d : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3282d7c429f447198d5be0085981232 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ec2a7394d544b8b883a378f30d91a0f : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb6509305c24441b672c698b8cae144 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd0264b83f0b4383b2669e2e0ed90429 : SUCCESS in 4m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2b108c5d70340e2b0a69e53b8e0e8eb : SUCCESS in 1h 37m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8176d28752764155b70981f9fe5e9c13 : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5dcf84eb1bfb7142370893f39cfa14caeabcdcdc","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2020-02-11 01:39:31.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for the work. I confrimed the change helps to work for stein.","accounts_in_message":[],"_revision_number":1},{"id":"5a635ef1aca4f9b1d30c603a0de5c7d79b5c425c","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2020-02-11 01:40:25.000000000","message":"Patch Set 1:\n\nBut we also have to upgrade keystoneauth version, do we have to bump the kesytoneauth\u0027s version up for stein?","accounts_in_message":[],"_revision_number":1},{"id":"453518ff5740647532d5f8c12a5a8f7c17a815b7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-11 09:33:58.000000000","message":"Patch Set 1:\n\n\u003e But we also have to upgrade keystoneauth version, do we have to\n \u003e bump the kesytoneauth\u0027s version up for stein?\n\nI suppose yes, as I see from releases page (https://releases.openstack.org/stein/index.html) it is possible, but let\u0027s see how the cores of keystoneauth like it :-)","accounts_in_message":[],"_revision_number":1},{"id":"01d9e758787d436b0a6e2c1ec33946bd12143937","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-02-11 20:36:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed85f0186e6f267b3bcc6a7211817c1de076fc65","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-12 11:37:20.000000000","message":"Patch Set 1: Workflow-1\n\nThe backport of keystoneauth patch is still under discussion, and it seems that we can\u0027t backport it.\nLet\u0027s wait the result of that discussion","accounts_in_message":[],"_revision_number":1},{"id":"1c6e701748be6460a6893cf76ce5fa0cce3d296f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 13:58:29.000000000","message":"Abandoned\n\nThe keystoneauth patch is blocked (I just abandoned it), if there\u0027s a real need for this fix, we can try ot workaround this in neutron-lib (that\u0027s ugly, but better than nothing)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"309c53a7d69f53cf0828559d70b7277dc7a07f98","revisions":{"309c53a7d69f53cf0828559d70b7277dc7a07f98":{"kind":"REWORK","_number":1,"created":"2020-02-10 13:00:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/20/706820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/706820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/706820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/706820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/706820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/706820/1"}}},"commit":{"parents":[{"commit":"3dcc921af53bc5012e3c89ff464084a8d0f3cf22","subject":"install neutron_lib international messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3dcc921af53bc5012e3c89ff464084a8d0f3cf22"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-05-31 09:50:46.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-02-10 13:00:45.000000000","tz":0},"subject":"placement client: fix routed prov networks working","message":"placement client: fix routed prov networks working\n\nRouted provider networks works with placement microversion 1.1, that\nversion returned no body for resource provider creation, but from 1.20\nbody is returned and the client expected that as bandwidth feature were\ndesigned after that.\n\nDepends-On: https://review.opendev.org/706774\nRelated-Bug: #1828543\nChange-Id: Id6e6d633b00237d8909160e7ed6f5e495399a252\n(cherry picked from commit 28e71cbd74bf10da3332265659adb27ee9614582)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/309c53a7d69f53cf0828559d70b7277dc7a07f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/309c53a7d69f53cf0828559d70b7277dc7a07f98"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
