)]}'
{"id":"openstack%2Fneutron-lib~708965","triplet_id":"openstack%2Fneutron-lib~master~I90419fe0b385e7bee216a52c1169aba6d6975d56","project":"openstack/neutron-lib","branch":"master","topic":"hacking2","hashtags":[],"change_id":"I90419fe0b385e7bee216a52c1169aba6d6975d56","subject":"Switch to hacking 2.0","status":"MERGED","created":"2020-02-20 21:09:29.000000000","updated":"2020-03-17 17:12:23.000000000","submitted":"2020-03-17 17:10:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708965-1584465041776-962e1dba","meta_rev_id":"86ef4c632050c4f5477093953cc385c06639a97a","_number":708965,"virtual_id_number":708965,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 17:10:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-03-17 08:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2020-03-17 11:03:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-03-17 15:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-03-17 10:15:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-03-17 14:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-03-17 15:36:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 08:58:50.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-03-17 10:15:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-03-17 11:03:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-17 14:02:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-17 15:36:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-17 17:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1bbef1b477779164d3987970e55988840e9ecf2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-20 21:09:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4974c5941666a5b5e3efc54a8f533d99acf60bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 22:40:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/19c2f20168b04499bacafde500912266 : SUCCESS in 5m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c6b9c14756248afa39654404fa46443 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1e6a4dfc91414d8b254a1dfc43d37e : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a397f340159742d89efef27f42f929d0 : FAILURE in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010bcc7fc9c3435a9f33ae5d561f8e5e : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d7441ff2be4b5db458ff9a95eb1b72 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e873b953c807493e8cc513fa059d7892 : FAILURE in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86be3cb9694d4ff6ae4b7ba0dcc43df7 : FAILURE in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80344b367b884a41bda7d666fc5b1ab8 : FAILURE in 6m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f9b9b402c5a4b56b1deb7ded7ab9139 : SUCCESS in 1h 29m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56f573b5d9a347fb8e31d6bbe37ea639 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"985798bd43f3b7177fbefe073d92a0928c64d8ec","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-21 08:00:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5095194b42eb7bf202d728d35e212b290ee85af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 09:21:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec999b064da6449882face3246419907 : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6edfb04cf5ad421ca656153249ae1359 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec78572cdb548328d38e06485462077 : SUCCESS in 7m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0cb453f225e41be89c9332800efe18b : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c0ebe2a2b16419d98422b2aff09a338 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab90d8d179d94f5b8a934da1dade8063 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/184afd8716d44db299e2e7d771e02246 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c2154319fc249f098aa123f5912b3fc : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc49dacc3f745f3ba63bf263269b9d6 : SUCCESS in 5m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/944891e656554e858bd43bf8217d0f16 : SUCCESS in 1h 17m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1857b02da86640478403760d0b2c3d08 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4754dbc7110abc5af03363aa0fe4059be0125610","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-02-24 13:57:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooking all good.","accounts_in_message":[],"_revision_number":2},{"id":"e32a90141edbf5e8cb23c0f88c88b903a668889f","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-02-24 14:11:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84138ac6dc9a3464e0cd741759825a5ef2350a89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-24 17:07:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1fbe868daed5ec14001f12d11f679ee158a39c0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-28 06:36:41.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"726921ba232ea6a81981fb854a51f8c14ab992a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-05 13:30:19.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Akihiro, I started to upload consumer patches for this with the same topic","accounts_in_message":[],"_revision_number":2},{"id":"5cd10675c9555aa3d8a660efc5fde8f9c982578c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-17 06:31:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f5c7b12bcaa5fc6eef8eb90e4c0149e7eecd72f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 08:08:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad0cc5426b1a48368b3d89a86bd7a24b : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f42f536604b49fbaf4a0b708c327456 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2977c6eb053c4e4e9d4fd675a4177dd2 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e16d5fada1db4e83ac91124ee7241a64 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/680685ffb7874036914c6a066e8ba621 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c45abfe59a940a28ada050f9b63ae10 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b89b4905e564317a39a538c1408839c : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/597c5c4134cb43e0a8f228c00677e438 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2df1e75a63a4f4db50f3c9da4f3b3f2 : SUCCESS in 11m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4808d0818284d8bb68d0ab8389f0188 : SUCCESS in 1h 28m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48807e822574458684815ee1f45763d7 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"c351c99f6bcd9f88446fbc0b2fbd229237660db8","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-03-17 08:58:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bf8adf1d247422b4c4bb86bfc0b63ff418bf03db","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-17 10:15:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"70acd74df677766cf66db1c9175cbb90729b3709","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-17 11:03:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"18f8d91d1349a97e425df096614272250dd36d57","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 14:02:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8103539c169854826334966818344770ccfa4d3d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-17 15:36:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5296377d18d13aa46257cee1916b92180b0a256c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 15:37:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7cf6b4174bd5bc58df9634c8a030e0ab3ce7f85c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:10:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e61ef140d489e9a8563397ee5a6c9236c1b4f91e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:10:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/519c71ecad7b4cfd9ff76e634d9105e8 : SUCCESS in 6m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7a8ee7e33814b2cadabb2a0456a64a0 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec2181ae32a54b75ad53a0848d6d2140 : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfd6720ae85b460b94426f8f6f99affc : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6efb89564ec142779e1f023acb4e74af : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fb141d2f51848fa82fce6485cb3dc60 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d199256182a4610a7861fac1450d91e : SUCCESS in 5m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/772f2ec04f0d449c9e32ea1a09662616 : SUCCESS in 1h 30m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f46d2ff160b94a7d9c76bc295fa78ab1 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"dd6173e28eeb90032d65e43f7f94928c597e7452","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:12:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a7c64fef5ee545da866a221996a0061b : SUCCESS in 1m 12s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3740fe1ab92d4656837680214bff3662 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f0666f7a72489cb97c0ce6e89a7b88 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4f5f1c7192e3f11bd514ce0e7f8959ef972e3be","revisions":{"6275ec3375e2c65649e3edf4df3ff6003b3b9ef9":{"kind":"REWORK","_number":1,"created":"2020-02-20 21:09:29.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/65/708965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/708965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/1"}}},"commit":{"parents":[{"commit":"c099a2278924f057445525030104b2cc6ac4d1e2","subject":"Update the minversion parameter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c099a2278924f057445525030104b2cc6ac4d1e2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-20 21:05:59.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-20 21:05:59.000000000","tz":540},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nIn hacking 2.0, local-check-factory was removed as it is not compatible\nwith flake8 3.x and it is advised to use flake8\u0027s local plugins [1].\nChecks specific to neutron-lib registered via local-check-factory are\nconverted into flake8\u0027s local plugins [2].\n\nNote that neutron_lib.hacking.checks.factory is kept not to break hacking\nchecks in neutron-lib consumers. They need to be converted into the style\nin hacking 2.x in each repository and then we can drop the factory here.\n\n[1] https://docs.openstack.org/releasenotes/hacking/unreleased.html#relnotes-2-0-0\n[2] https://flake8.pycqa.org/en/3.7.0/user/configuration.html#using-local-plugins\n\nChange-Id: I90419fe0b385e7bee216a52c1169aba6d6975d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6275ec3375e2c65649e3edf4df3ff6003b3b9ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6275ec3375e2c65649e3edf4df3ff6003b3b9ef9"}]},"branch":"refs/heads/master"},"27868e161a4b10e6c3ab29c82872aa08359f5a60":{"kind":"REWORK","_number":2,"created":"2020-02-21 08:00:04.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/65/708965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/708965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/2"}}},"commit":{"parents":[{"commit":"c099a2278924f057445525030104b2cc6ac4d1e2","subject":"Update the minversion parameter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c099a2278924f057445525030104b2cc6ac4d1e2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-20 21:05:59.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-21 07:59:43.000000000","tz":540},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nIn hacking 2.0, local-check-factory was removed as it is not compatible\nwith flake8 3.x and it is advised to use flake8\u0027s local plugins [1].\nChecks specific to neutron-lib registered via local-check-factory are\nconverted into flake8\u0027s local plugins [2].\n\nNote that neutron_lib.hacking.checks.factory is kept not to break hacking\nchecks in neutron-lib consumers. They need to be converted into the style\nin hacking 2.x in each repository and then we can drop the factory here.\n\n[1] https://docs.openstack.org/releasenotes/hacking/unreleased.html#relnotes-2-0-0\n[2] https://flake8.pycqa.org/en/3.7.0/user/configuration.html#using-local-plugins\n\nChange-Id: I90419fe0b385e7bee216a52c1169aba6d6975d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27868e161a4b10e6c3ab29c82872aa08359f5a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/27868e161a4b10e6c3ab29c82872aa08359f5a60"}]},"branch":"refs/heads/master"},"f4f5f1c7192e3f11bd514ce0e7f8959ef972e3be":{"kind":"REWORK","_number":3,"created":"2020-03-17 06:31:44.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/65/708965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/708965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/708965/3"}}},"commit":{"parents":[{"commit":"c099a2278924f057445525030104b2cc6ac4d1e2","subject":"Update the minversion parameter.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c099a2278924f057445525030104b2cc6ac4d1e2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-20 21:05:59.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-17 06:31:31.000000000","tz":540},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nIn hacking 2.0, local-check-factory was removed as it is not compatible\nwith flake8 3.x and it is advised to use flake8\u0027s local plugins [1].\nChecks specific to neutron-lib registered via local-check-factory are\nconverted into flake8\u0027s local plugins [2].\n\nNote that neutron_lib.hacking.checks.factory is kept not to break hacking\nchecks in neutron-lib consumers. They need to be converted into the style\nin hacking 2.x in each repository and then we can drop the factory here.\n\n[1] https://docs.openstack.org/releasenotes/hacking/unreleased.html#relnotes-2-0-0\n[2] https://flake8.pycqa.org/en/3.7.0/user/configuration.html#using-local-plugins\n\nChange-Id: I90419fe0b385e7bee216a52c1169aba6d6975d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f4f5f1c7192e3f11bd514ce0e7f8959ef972e3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f4f5f1c7192e3f11bd514ce0e7f8959ef972e3be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
