)]}'
{"id":"openstack%2Fneutron-lib~725815","triplet_id":"openstack%2Fneutron-lib~master~Ia005547d7a8353b629f2262f424d7761c45b185f","project":"openstack/neutron-lib","branch":"master","topic":"hacking","hashtags":[],"change_id":"Ia005547d7a8353b629f2262f424d7761c45b185f","subject":"Update to hacking 3.0","status":"MERGED","created":"2020-05-06 10:01:36.000000000","updated":"2020-05-06 22:56:31.000000000","submitted":"2020-05-06 22:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725815-1588805689301-92884968","meta_rev_id":"b12b2d5f6beb475a01b8f40ba9705d438bf93c52","_number":725815,"virtual_id_number":725815,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 22:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-05-06 10:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-05-06 21:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-05-06 21:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 10:25:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-06 21:17:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-06 22:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13b8a85f5a305030cd390f259ce1b6dbe1e53115","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 10:01:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99d7d77351ab581dc16a482d4f441221b8fab2b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-06 10:25:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c0990dd694b219b74a884c6496487ede0cb5636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0383de256eee43618cd66044f2f2e0a7 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c1ca14ecb584c66ab3070a38c2ac0a6 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea952948a7824254adcebc36a43ffedd : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12bd5b652e93411b966356027fe9560f : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e88556fe9954cfcbe5e2843de076b78 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4010a82a4343e2aafa1fb968eafe09 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32bc696723d547ae8bb21b750b2c08cd : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f30bf8da7af4ce5bae5b5635046d9b3 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fb9fcb85bdc4f5988efba3cfc58ed44 : SUCCESS in 5m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8953c96c210e49e8865c7d7d7e9ba71f : SUCCESS in 1h 59m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15bebbb2d6b447338d51bb0ae6607100 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ab114059fdbf0dab099bff6424bf0d4c79af5d26","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-06 21:17:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"947c44e7c754c7abe3cf687c904162865dc94e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 21:18:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06cafcec32888b1fb525d86a1b667c6d8f2f9634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:54:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0c3ec30f4da94fc4b731f0d0bfababa1 : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfd27c2c2da4463e82c283d70b69fd78 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7877b206bb49959aca340d93e820a2 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc502ff5881544f8b86f7dd1abebc2b3 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f751b7089dce483b9dfd8941e62bb604 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1ce6fede0594208849af52b564a3d2c : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/486e8c19c08b47f79c8b75d7ae334971 : SUCCESS in 5m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82f1853ae4304a5a81d401825b73c35c : SUCCESS in 1h 34m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212cf444584e4445b4ce8c21df5690e7 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9a7dbfabbe16c2635e894aecd403f922b2f809a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:54:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"04cfe32150dda8fb9772efbd457170aa574144e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d2f7a328e04c4e28a01d3be93e253d25 : SUCCESS in 1m 21s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a76052402a64e16bbf385fb0e9eb3ab : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e369c11dbefe446089286e19ae84fe64 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a585ca069b1089bacedb32d254bc8066afd15211","revisions":{"a585ca069b1089bacedb32d254bc8066afd15211":{"kind":"REWORK","_number":1,"created":"2020-05-06 10:01:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/725815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/15/725815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/725815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/725815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/725815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/15/725815/1"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-06 10:00:23.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-06 10:00:23.000000000","tz":120},"subject":"Update to hacking 3.0","message":"Update to hacking 3.0\n\nUpdate to newest hacking version, it has only minimal changes and full\nsupport for Python 3.\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nChange-Id: Ia005547d7a8353b629f2262f424d7761c45b185f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a585ca069b1089bacedb32d254bc8066afd15211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a585ca069b1089bacedb32d254bc8066afd15211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
