)]}'
{"id":"openstack%2Fneutron-lib~726053","triplet_id":"openstack%2Fneutron-lib~stable%2Fussuri~I0dfb076c342068c2fbddbc60941a14031ccc0e7f","project":"openstack/neutron-lib","branch":"stable/ussuri","topic":"bug/1836263","hashtags":[],"change_id":"I0dfb076c342068c2fbddbc60941a14031ccc0e7f","subject":"Documentation update - sending null in put update will be considered as {}. Closes-Bug: #1836263","status":"ABANDONED","created":"2020-05-07 07:31:54.000000000","updated":"2020-05-13 09:49:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5e37100751d0bfb3bda2a575e2a012539d1ac7c","_number":726053,"virtual_id_number":726053,"owner":{"_account_id":18044,"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","username":"mahajan-abhishek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-07 07:47:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":-1,"date":"2020-05-07 09:15:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 07:47:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-07 09:15:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"0607a910062438a0261b745785484ff7f3fa098c","author":{"_account_id":18044,"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","username":"mahajan-abhishek"},"date":"2020-05-07 07:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b3ad1b9ce5f990a6dabea0285e868aa53198b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 07:47:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412e26dbf8044255b861add873993950 : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae68e2d200a4e46820808c9707745d4 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2b927268c9c479462ef89f1523180066f14381ac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-07 09:15:43.000000000","message":"Patch Set 1: Code-Review-1\n\nWhy not on master?","accounts_in_message":[],"_revision_number":1},{"id":"1091886f6fc43d19cdd9dbcb73b04f120890f324","author":{"_account_id":18044,"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","username":"mahajan-abhishek"},"date":"2020-05-13 09:49:11.000000000","message":"Abandoned\n\nFix proposed in following review: https://review.opendev.org/726293","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9419d32c4e82da0f5dec01c7bfa90cac521c19d","revisions":{"b9419d32c4e82da0f5dec01c7bfa90cac521c19d":{"kind":"REWORK","_number":1,"created":"2020-05-07 07:31:54.000000000","uploader":{"_account_id":18044,"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","username":"mahajan-abhishek"},"ref":"refs/changes/53/726053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/53/726053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/726053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/726053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/53/726053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/53/726053/1"}}},"commit":{"parents":[{"commit":"a1528e7f943d3bde3883d58ec30ca8b13cff4592","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a1528e7f943d3bde3883d58ec30ca8b13cff4592"}]}],"author":{"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","date":"2020-05-07 07:15:46.000000000","tz":330},"committer":{"name":"Abhishek Mahajan","email":"m.abhishek84@gmail.com","date":"2020-05-07 07:31:07.000000000","tz":330},"subject":"Documentation update - sending null in put update will be considered as {}. Closes-Bug: #1836263","message":"Documentation update - sending null in put update will be considered as {}.\nCloses-Bug: #1836263\n\nChange-Id: I0dfb076c342068c2fbddbc60941a14031ccc0e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9419d32c4e82da0f5dec01c7bfa90cac521c19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9419d32c4e82da0f5dec01c7bfa90cac521c19d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
