)]}'
{"id":"openstack%2Fneutron-lib~726293","triplet_id":"openstack%2Fneutron-lib~master~I5fe45af8a8075e52d3a196b24fc926343018d8fd","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I5fe45af8a8075e52d3a196b24fc926343018d8fd","subject":"[Docs] Note about port\u0027s binding_profile update","status":"MERGED","created":"2020-05-08 07:08:01.000000000","updated":"2020-06-15 08:01:42.000000000","submitted":"2020-06-15 07:59:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726293-1592207985540-2c5ff2b2","meta_rev_id":"92bea30e505c101f611df4f68b83940b70720543","_number":726293,"virtual_id_number":726293,"owner":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 07:59:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-06-11 12:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-06-11 12:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-06-11 13:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-15 07:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 06:38:55.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-06-11 12:29:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-11 13:28:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-06-15 07:53:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-15 07:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30b9986e7b51d0d6dc7aa4ce5796b673723cc825","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-05-08 07:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d89e568547f4e72942452090d74859da633b3785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 07:21:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6f2ee32bfc74322a5c4be6e24b99d13 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c4be1314414becbd851112c36c6caa : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8af2d1eea5d42b5b6230c38e9c7a435 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"615f9aa67af30f90c05300c3957b73a7bc22a007","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-08 07:31:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08233a0d603e23985696dc6c5ab7fbe8a7473a80","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-05-08 08:39:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ad1d71ed4f4181c225c347d4137a9f1cee389b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 08:49:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/535de2db5af74bdea4c8f68837e72e2e : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a26643ab4c4963bbcc4bebd0032aea : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbbe385da1e4d5d9aa3cd0205b7ace9 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9c5eed50563ecab79deb7bf861d7dab58096db6a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-08 08:54:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e4e47242043e7e002ac1a689d708010b41cfaf5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-08 10:08:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nHi. Thx for Your contribution to the neutron-lib :)","accounts_in_message":[],"_revision_number":2},{"id":"04ee50c9fcfb8207077157086b623ee487c0ec81","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-05-11 07:37:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b336513764fe7fbdca1829c42d4079185f202fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 07:56:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9434ad7d86604c198d51d7edb3f95022 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9598261ad7a4919a8aff4a53f6cec77 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf496bb9e3954caca0abca4356b310b4 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"de3aa229193e02198811cacd56890b976af9cc52","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-11 12:45:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f329de20163714cdce2c3a6eadc4b17b1c342c6","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-05-15 04:56:13.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nMay I know what changes are required to close the bug.","accounts_in_message":[],"_revision_number":3},{"id":"4bc108d8f09fd083b78fc7c0cdf96cbaafc5bfd3","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-06-03 13:10:53.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0c74ce6a581e67b7be11a3ad3f94c1988eee273a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 13:17:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d51c844df9d7497ba25950a441e7ed16 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8257cd11d81466e9bc52a97bb026f13 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b56f9d365049ac951319839268b7f5 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"59515508869fcf98811458e3197466886a916cea","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-06-04 06:38:55.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d6e3eb0207d77ee9202e3158d3456d017bdbdb36","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-05 09:11:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThanks for working on this, and sorry for late reply, my suggestion is to add some informative sentence to parametrs.yaml where binding:profile field description is","accounts_in_message":[],"_revision_number":4},{"id":"8f0cd0c097ae49e57b082a41c5937f7727afbb36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-05 09:27:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2bd3d255ffcb6ac3cfd312f65940a55159b72de","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-06-09 07:04:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"72f0437b02e9e306c0d5bf4aadf5b1e96a96129a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 07:23:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d481a8e0fee4731abbd049b7c5aba26 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9423c9b80cb5492a8b8f6c47025d7625 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371e3fa85a224cd0aa90e7063849f7e0 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f30da71f483badc287db72f0f9b18bef8715dcd6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-09 14:07:59.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nLooks good as I see, please remove the extra whitespace here: https://review.opendev.org/#/c/726293/5/api-ref/source/v2/parameters.yaml@1961","accounts_in_message":[],"_revision_number":5},{"id":"436173673ccc9c4fdbd5160888ad191987df9e69","author":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"date":"2020-06-09 19:11:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7ec14076098f2e0527aaab88f517615079a30b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:26:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/69d1685d5baa497b93882be1f03dde41 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f91de999f2400781fb75aaba56d4a9 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1426f706706f4ed0b2720db725f8968d : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":6},{"id":"f434a8add328ff2e87708dae55571b870f510feb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-11 12:02:17.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7bbb2a7d5b19ff330b72f1a6845661b8fac69375","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-11 12:02:21.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"4468c3b9aa1cad4a344e38f005863b6f3433a343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:11:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0aa3992be16744379835efa421e2cf03 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb43048b7f6e4422960e4212c47ec8a8 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae71fd6dafb84928aa14bec74d87092c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":7},{"id":"a698b29a9d4b9b97ae64211de3ef1c7850ebaea7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-11 12:29:58.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"06db46740f9c313af3eab31f8c3ba4f0473be052","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-11 13:28:22.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1ac560586d1a53eab4eaf0815fca8f0f91018a28","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-15 07:53:59.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"33edcc8b1a7f0267d21eb3cf810efc01ff06b859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:54:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2814783d6dc220849278db40d0ec5248d4dab2ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:59:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b1a79d892d042b99bc4454c666487bf : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ecae1696b1480a930ed6769575813c : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2088ebab954aa2b6c728ff7c546b12 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":7},{"id":"67e29a50e9a369b1c87ef8751acb856608ff3b19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:59:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"92bea30e505c101f611df4f68b83940b70720543","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 08:01:42.000000000","message":"Patch Set 7:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f5305c7cf8de4785a0553768bb187977 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab2607063a54ff1838a026c361bcbde : FAILURE in 1m 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4c2e5ce0d043e5ca8ac1120f961cd1eea9f9e04b","revisions":{"b2987cf842268905a59b3f26e2b6260236b1e255":{"kind":"REWORK","_number":1,"created":"2020-05-08 07:08:01.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/1"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-08 07:07:55.000000000","tz":330},"subject":"This is a bug fix","message":"This is a bug fix\n\nthis is the documentation bug in neutron api-ref\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2987cf842268905a59b3f26e2b6260236b1e255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2987cf842268905a59b3f26e2b6260236b1e255"}]},"branch":"refs/heads/master"},"f5c3255318c807fcc5bcc1cd062abe02a6727020":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-08 08:39:36.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/2"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-08 08:26:13.000000000","tz":330},"subject":"This is a fix proposed for neutron api-ref docs","message":"This is a fix proposed for neutron api-ref docs\n\nIf any resource of port is updated to null using the put command then it\nis considered as {} in the response which should be documented is the\nbug,so it is documented in neutron api-ref.\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5c3255318c807fcc5bcc1cd062abe02a6727020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5c3255318c807fcc5bcc1cd062abe02a6727020"}]},"branch":"refs/heads/master"},"f6904d92e1fa83ece1324c658d5c9ea2174e181d":{"kind":"REWORK","_number":3,"created":"2020-05-11 07:37:19.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/3"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-11 07:33:34.000000000","tz":330},"subject":"[Docs] Note about port\u0027s binding_profile update","message":"[Docs] Note about port\u0027s binding_profile update\n\nThe fix is to document that, if binding profile of a port is updated to\nnull using put command then it is treated as {} in response.\n\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f6904d92e1fa83ece1324c658d5c9ea2174e181d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f6904d92e1fa83ece1324c658d5c9ea2174e181d"}]},"branch":"refs/heads/master"},"5f20be50ebcb4b11728a558330b0e079480a6def":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-03 13:10:53.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/4"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-06-03 13:09:30.000000000","tz":330},"subject":"[Docs] Note about port\u0027s binding_profile update","message":"[Docs] Note about port\u0027s binding_profile update\n\nThe fix is to document that, if binding profile of a port is updated to\nnull using put command then it is treated as {} in response.\n\nPartial-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5f20be50ebcb4b11728a558330b0e079480a6def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5f20be50ebcb4b11728a558330b0e079480a6def"}]},"branch":"refs/heads/master"},"bdf2d43c9067d01c455578e87f97535def3af0b1":{"kind":"REWORK","_number":5,"created":"2020-06-09 07:04:12.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/5"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-06-09 07:03:00.000000000","tz":330},"subject":"[Docs] Note about port\u0027s binding_profile update","message":"[Docs] Note about port\u0027s binding_profile update\n\nThe fix is to document that, if binding profile of a port is updated to\nnull using put command then it is treated as {} in response.\n\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdf2d43c9067d01c455578e87f97535def3af0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bdf2d43c9067d01c455578e87f97535def3af0b1"}]},"branch":"refs/heads/master"},"49ed645dff69122bb309c77d034a8d7290bef039":{"kind":"REWORK","_number":6,"created":"2020-06-09 19:11:48.000000000","uploader":{"_account_id":30907,"name":"Triveni Gurram","email":"triveni.gurram@tcs.com","username":"triveni"},"ref":"refs/changes/93/726293/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/6"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-06-09 19:10:48.000000000","tz":330},"subject":"[Docs] Note about port\u0027s binding_profile update","message":"[Docs] Note about port\u0027s binding_profile update\n\nThe fix is to document that, if binding profile of a port is updated to\nnull using put command then it is treated as {} in response.\n\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/49ed645dff69122bb309c77d034a8d7290bef039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/49ed645dff69122bb309c77d034a8d7290bef039"}]},"branch":"refs/heads/master"},"4c2e5ce0d043e5ca8ac1120f961cd1eea9f9e04b":{"kind":"REWORK","_number":7,"created":"2020-06-11 12:02:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/726293/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/93/726293/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/93/726293/7"}}},"commit":{"parents":[{"commit":"cc11734bbc3628f7225c61ec4d1f1ed21660d912","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cc11734bbc3628f7225c61ec4d1f1ed21660d912"}]}],"author":{"name":"triveni","email":"triveni.gurram@tcs.com","date":"2020-05-06 02:36:19.000000000","tz":330},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-11 12:02:11.000000000","tz":0},"subject":"[Docs] Note about port\u0027s binding_profile update","message":"[Docs] Note about port\u0027s binding_profile update\n\nThe fix is to document that, if binding profile of a port is updated to\nnull using put command then it is treated as {} in response.\n\nCloses-Bug: #1836263\n\nChange-Id: I5fe45af8a8075e52d3a196b24fc926343018d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4c2e5ce0d043e5ca8ac1120f961cd1eea9f9e04b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4c2e5ce0d043e5ca8ac1120f961cd1eea9f9e04b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
