)]}'
{"id":"openstack%2Fneutron-lib~727552","triplet_id":"openstack%2Fneutron-lib~master~Ifb13eed591c7ada905348d6ea86fb22dd73908bb","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"Ifb13eed591c7ada905348d6ea86fb22dd73908bb","subject":"Fix hacking min version to 3.0.1","status":"ABANDONED","created":"2020-05-13 00:54:28.000000000","updated":"2020-05-15 01:39:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32b7c60e1528a753eb167e54c49aba1d27e3c1cf","_number":727552,"virtual_id_number":727552,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-13 02:38:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":-1,"date":"2020-05-13 06:37:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 02:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-13 06:37:51.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"df6c7f3835ef86b60d12ea641b1fbe0e3c0bb70c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 00:54:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca05dec49d50ce32c7c33a46e074fbfd8e0fb27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:38:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/06222a852da54f5e8ee30a93cd0beb85 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f73bb6a870d4d0aaedf9a279cf2a960 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6e75b796f94f50be41f7383033d749 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0d447cb53344f297091b2cbed1a179 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fed817ed5fb4b05b2f51fc5222954b7 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdaaf990163457fb2ba567475b3e65c : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a8f18aabb884d248bd1be375582ec1c : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/974ee98c7d1c41e7b8d73c75489f95e4 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4602e8a041154de6baf36492d9d1ae6d : SUCCESS in 5m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a05725704af54c878d38979b15ac1aed : SUCCESS in 1h 32m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c41a0274a0fa4ca7bba0ae42081acd00 : SUCCESS in 5m 00s\n\nWarning:\n  Comments left for invalid file neutron_lib/utils/helpers.py","accounts_in_message":[],"_revision_number":1},{"id":"9da5c19f434bb9a08653af321804cb57e1490c1f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-13 06:37:51.000000000","message":"Patch Set 1: Code-Review-1\n\nhttps://review.opendev.org/#/c/727470/ was first","accounts_in_message":[],"_revision_number":1},{"id":"32b7c60e1528a753eb167e54c49aba1d27e3c1cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-15 01:39:38.000000000","message":"Abandoned\n\nfixed in https://review.opendev.org/#/c/727470/1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb44e56ca9db87366f6683fa0e7c4e5eb77da9d8","revisions":{"cb44e56ca9db87366f6683fa0e7c4e5eb77da9d8":{"kind":"REWORK","_number":1,"created":"2020-05-13 00:54:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/52/727552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/727552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/727552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/727552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/727552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/727552/1"}}},"commit":{"parents":[{"commit":"a585ca069b1089bacedb32d254bc8066afd15211","subject":"Update to hacking 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a585ca069b1089bacedb32d254bc8066afd15211"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 00:53:56.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 00:54:27.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: Ifb13eed591c7ada905348d6ea86fb22dd73908bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb44e56ca9db87366f6683fa0e7c4e5eb77da9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb44e56ca9db87366f6683fa0e7c4e5eb77da9d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
