)]}'
{"id":"openstack%2Fneutron-lib~729615","triplet_id":"openstack%2Fneutron-lib~master~I3526d62d8ea9f6d0eadab88472347b35117c7dd4","project":"openstack/neutron-lib","branch":"master","topic":"bug/1879716","hashtags":[],"change_id":"I3526d62d8ea9f6d0eadab88472347b35117c7dd4","subject":"Fix policy enforcement for network mtu","status":"MERGED","created":"2020-05-20 14:20:23.000000000","updated":"2020-06-19 00:20:55.000000000","submitted":"2020-06-19 00:19:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729615-1592525964851-b38bd9c1","meta_rev_id":"4a82be34dd625031b94804d63443b9400f2e1780","_number":729615,"virtual_id_number":729615,"owner":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-05-21 07:02:36.000000000","_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 00:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"all":[{"value":1,"date":"2020-05-20 15:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"value":2,"date":"2020-06-18 22:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-05-22 08:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-05-21 10:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-05-20 18:34:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},{"value":1,"date":"2020-05-20 15:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":1,"date":"2020-05-21 07:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-20 16:22:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"value":1,"date":"2020-06-18 22:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"date":"2020-05-27 09:34:30.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-05-22 08:26:33.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},{"value":0,"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},{"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 15:31:40.000000000","updated_by":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"reviewer":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"state":"REVIEWER"},{"updated":"2020-05-20 15:40:52.000000000","updated_by":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"reviewer":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"state":"REVIEWER"},{"updated":"2020-05-20 16:22:37.000000000","updated_by":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"reviewer":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"state":"REVIEWER"},{"updated":"2020-05-20 18:34:34.000000000","updated_by":{"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},"reviewer":{"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},"state":"REVIEWER"},{"updated":"2020-05-21 07:57:13.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-05-21 10:33:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-22 08:26:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-22 08:26:33.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-27 09:34:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-18 22:54:17.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-06-19 00:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4188459f5f40daa337293612424ee650fe6834c0","author":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"date":"2020-05-20 14:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"950591f0d334596abb3926a7ebd15c7382127fef","author":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"date":"2020-05-20 14:21:04.000000000","message":"Topic set to bug/1879716","accounts_in_message":[],"_revision_number":1},{"id":"f4f3127a3d3890329703b2d050b7248179b4fff9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-20 14:50:24.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease file a bug (if not exist) and add it to the commit message as Closes-Bug: #\u003cbug_id\u003e","accounts_in_message":[],"_revision_number":1},{"id":"ad1bfaff8dc243ce83c506c0cdd72cd76b7a673d","author":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"date":"2020-05-20 14:51:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d8d6bfe57f537467ec14f23dd4452c66c8c7ac7c","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-05-20 15:31:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7bae8a0c29e7de21888978fee80ecaf757f14ff1","author":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"date":"2020-05-20 15:40:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"797fdc321831ee2df8da95e4282e895a0468da1d","author":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"date":"2020-05-20 16:22:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bca023b569af2e9596f166e932de866d35029ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 16:31:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c6c4a318954473a5e22cc13da4d496 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c02b2536b9b44b018e5350dbba634754 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/522210005dd2440ebddd33c9d8e247eb : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fde98e757c4713a6f0cdd961be5502 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c522f3c9cc8d4bbf896e744c1158a33c : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d949e764284924b7ca34c5ca22ee7f : SUCCESS in 6m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54c09dd4d4e14f82b98cbb8d9d015334 : POST_FAILURE in 1h 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"41174e04f4835a8706110453fde9f4f6b47831da","author":{"_account_id":30746,"name":"Travis Neely","email":"travis.neely@att.com","username":"travis.neely@att.com"},"date":"2020-05-20 18:34:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7486244b9bfbb57c4410f62991abf95aaf0ef5d3","author":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"date":"2020-05-21 07:02:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d06484365952cb9e33f095826b972a456069f05a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-21 07:57:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf3375a6ef1b9cdbc8c0eb54a480c86cb5357b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 08:30:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cbd9c2c17d445ebdd23ade73af4dfb : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/454bc5bf802d4fad955d5e54ff6f188a : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74799556e86d435ca49b68ba0b1a473a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049417de8c0a41af88cab7bc297adfd9 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb0e074f77f4d3597a9983d3c1db38e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/552555ac2d7f49ecb0deb90608dd89d3 : SUCCESS in 5m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3863be0930974f72baf85d236c1f17ca : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"59b55b27dbbbca096573a95df757e654be7bc5fd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-21 10:33:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35df547054070278c9682138f91c89578ceeb5ce","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-22 08:26:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nBefore approving this, I would like Akihiro to take a look at this change as he is our policy and API expert.","accounts_in_message":[],"_revision_number":2},{"id":"5778ba4b7d829089819a10e452e186073c8ac580","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-06-18 22:54:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"281b5d40adf9d5be5b9041f52a7a29f0769a91ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:54:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d31ec495ef6a29a54d99e456a29a6945603530f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:19:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb22440d02404e958ba29c9b9a23b874 : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e82d8a8c6fa420fbbde5aa219f1cc13 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2d7a4195e14f48b9fae05348cd480f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3444df575e41508b806de252c96097 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b815b6ce42c4664b05af498b7466527 : SUCCESS in 5m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1bed235794d4b2381efc75212101bb4 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e9b64772c88b4d039ab60692e99823989483e8b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:19:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4a82be34dd625031b94804d63443b9400f2e1780","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:20:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83198f2024014b1fa97f50f74fe271f8 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfeee711840ff7a15e4fe0a527225d685ae690fc","revisions":{"39a2f880bd76fc551547f71d52fc604697515fd1":{"kind":"REWORK","_number":1,"created":"2020-05-20 14:20:23.000000000","uploader":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"ref":"refs/changes/15/729615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/15/729615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/1"}}},"commit":{"parents":[{"commit":"37b778fbc2a0b861403e34972c703381d756df90","subject":"Merge \"Add aggressive negotiation mode for ikepolicy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37b778fbc2a0b861403e34972c703381d756df90"}]}],"author":{"name":"rajesh.kudaka","email":"rajesh.kudaka@imaginea.com","date":"2020-05-20 14:18:45.000000000","tz":-300},"committer":{"name":"rajesh.kudaka","email":"rajesh.kudaka@imaginea.com","date":"2020-05-20 14:20:19.000000000","tz":-300},"subject":"Fix policy enforcement for network mtu","message":"Fix policy enforcement for network mtu\n\nChange-Id: I3526d62d8ea9f6d0eadab88472347b35117c7dd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/39a2f880bd76fc551547f71d52fc604697515fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/39a2f880bd76fc551547f71d52fc604697515fd1"}]},"branch":"refs/heads/master"},"cfeee711840ff7a15e4fe0a527225d685ae690fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-20 14:51:46.000000000","uploader":{"_account_id":29397,"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","username":"rajesh.kudaka"},"ref":"refs/changes/15/729615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/15/729615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/15/729615/2"}}},"commit":{"parents":[{"commit":"37b778fbc2a0b861403e34972c703381d756df90","subject":"Merge \"Add aggressive negotiation mode for ikepolicy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/37b778fbc2a0b861403e34972c703381d756df90"}]}],"author":{"name":"rajesh.kudaka","email":"rajesh.kudaka@imaginea.com","date":"2020-05-20 14:18:45.000000000","tz":-300},"committer":{"name":"Kudaka Poorna Rajesh","email":"rajesh.kudaka@imaginea.com","date":"2020-05-20 14:51:26.000000000","tz":0},"subject":"Fix policy enforcement for network mtu","message":"Fix policy enforcement for network mtu\n\nChange-Id: I3526d62d8ea9f6d0eadab88472347b35117c7dd4\nCloses-Bug: #1879716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cfeee711840ff7a15e4fe0a527225d685ae690fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cfeee711840ff7a15e4fe0a527225d685ae690fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
