)]}'
{"id":"openstack%2Fneutron-lib~730736","triplet_id":"openstack%2Fneutron-lib~master~I3f303583e8b7c56a2759453403eb73940e9ae089","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I3f303583e8b7c56a2759453403eb73940e9ae089","subject":"Add `access_as_readonly` sharing action","status":"MERGED","created":"2020-05-26 09:09:14.000000000","updated":"2021-01-08 13:19:48.000000000","submitted":"2021-01-08 13:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730736","meta_rev_id":"164da362f3f6fc35b6188a06882215a06c28a233","_number":730736,"virtual_id_number":730736,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 13:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-08 11:45:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-11-19 08:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-08 11:45:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-09-03 14:20:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 09:49:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-09-03 14:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 14:20:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-19 08:12:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d580744403a70ad8b5241f33fee1167b008ee40e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-26 09:09:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"394f391ebe9129e698af1e9193fd311fd1d74d27","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-26 09:10:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fb90f913432cd65a34bbbad8143fcf95cd87e976","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 09:49:38.000000000","message":"Patch Set 1:\n\nI\u0027m ok with this patch but adding a release note about those three constants, specially \"access_as_readonly\", to briefly document what does it mean.","accounts_in_message":[],"_revision_number":1},{"id":"56854083cc3437529502c47f38506726ab1629ce","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-26 10:24:27.000000000","message":"Patch Set 1:\n\nThanks for the review! :)\nDefinitely will add a release note before I\u0027m done, this is still very WIP.","accounts_in_message":[],"_revision_number":1},{"id":"8ea953c2180f245fde4555787d340b76209725c7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-26 10:28:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1549b3f5259964804c60c001958c7462b7fdfcf0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-26 10:44:54.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"dc5d014f5317002f122e7654483aa385eff478f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 11:47:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a0f458291841dfb866f40af8e36ab1 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7f23eec1f94f80899d6f44ee93c79e : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4079ae98a2b4b4e85a0639325092cdf : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c51ed28f2ac4e5092bf3aedbbf84a75 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b59fe5db424215883d413a21acd248 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e09d59767a24e9b94a427f4d2374680 : SUCCESS in 5m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/451333e6684345a4abb03f32b8c81ce8 : SUCCESS in 1h 17m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92f8bb723f324270a60ab5f4fb6aa0c5 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"07f0da9616db91bbb448ea3d2ed83c0dea8f02b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-25 20:43:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce40934a129654113bb371d17e005a748654cecf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-25 20:44:09.000000000","message":"Patch Set 3:\n\nRebased to fix merge conflict.\n\nAdam - was there a reason you had set -W ?","accounts_in_message":[],"_revision_number":3},{"id":"4b97008824599de0aed4316ad27ad41857bca993","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-25 20:46:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dce64e920c4a3bab3cf4415eee32f5d01639246e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 21:41:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d25bd1a13fe4337b28f092db4056ff6 : SUCCESS in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8003018b3ff44edea78457a113e848bc : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c24b5d2a8ec845c2a38495fbc7568273 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d977fe9845e649b7b965fde71ff6b47b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bedc08621014844b042129d522f8293 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7ba43a49774f0ab146d864265a248d : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fec002c795d4d8994faed23d496e9e7 : SUCCESS in 56m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a59414699c940c39d79cfd9c498e9ba : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3},{"id":"c46f1a633a2270a6c6a37fabb39a1d326d82cd7e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 12:56:01.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1c63831dd088d7700c1f1fa03454d6b0c4b1f9a9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 12:57:13.000000000","message":"Patch Set 4: Workflow-1\n\nRemoved the WIP from commit message, let me ping Adam to see if we can merge this","accounts_in_message":[],"_revision_number":4},{"id":"1b5b18bcd3f559d4c1e7bcba3a4c7c135a7b2eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 14:18:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d31d8fb74674b88aeb846881cadeac4 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e037b9e2e7c94db5ad273bc657c6be1b : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8360255d82674d00ac67b505d844100f : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a72b053c6d243c18ea39f38a4c9bf3d : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7887d6f84f2d4b20a39af5ee9cef99c5 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b39ebe76ddeb4c0482c9dd5532a89375 : SUCCESS in 7m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbf94c337d4f48c9b6f8c1455a8e1db4 : SUCCESS in 1h 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70bad1f2475d43f19d9f856ae7579a7b : SUCCESS in 4m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba21bf6ae8e84fba8acade87670ca2ed : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":4},{"id":"436aebf0cf836469fac5a5c4ecef4117c98954f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 14:20:57.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"11e00c35e0959e01edad40c2e1dfaa9004a33994","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-19 08:12:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9b763aa4c238b05fb4ad81f835b81e91130e116d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 17:36:18.000000000","message":"Patch Set 4:\n\nJust a question: how does this feature fit with the secure RBAC refactor? In particular with SGs [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/765394/4/neutron/conf/policies/security_group.py","accounts_in_message":[],"_revision_number":4},{"id":"7516d2a5b0139ac000877bd06bd27c3dd896739c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-08 11:45:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"85710fb35ed9b8e4bebf48480cc6fa72695f86ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 11:45:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ffd4c44175b9e6318324421de6bce9326051069f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:18:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80fa32ece1fe4caa9c46fac6e18836e6 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4144404d0064be8853f49059eb99356 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0d13eebdd94c8188cd2a047207306d : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec6ff20ea3ac450c8b172a66854f8816 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0543fd4e08e6492c94285125e85e006d : SUCCESS in 5m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb51872266594f04b477f00e0b12f5d7 : SUCCESS in 1h 29m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cde9e978cbbf42cebabcdff7b4d76929 : SUCCESS in 4m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6a49cf2a2c88453f95b7ea9fc119fd0c : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":4},{"id":"4b3acf9b1697500e96ba9cf8da76e2917b8795de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:18:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"164da362f3f6fc35b6188a06882215a06c28a233","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:19:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab0f64c9fc44741abbe05dd0494658e : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/219b390fcb074fa992c5f06910290c30 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"83019b98b6925998404f0bf37177dac3ff0cd3be","revisions":{"ff6f45588393d3cbd3c63a79bad97b021b380354":{"kind":"REWORK","_number":1,"created":"2020-05-26 09:09:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/730736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/730736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/1"}}},"commit":{"parents":[{"commit":"9e48026b1ba04a7bcd04f1083114b497e9199016","subject":"Merge \"Remove redefined constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e48026b1ba04a7bcd04f1083114b497e9199016"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 09:03:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 09:09:12.000000000","tz":-420},"subject":"WIP: Add `access_as_readonly` sharing action","message":"WIP: Add `access_as_readonly` sharing action\n\nChange-Id: I3f303583e8b7c56a2759453403eb73940e9ae089\nPartial-Bug: 1875516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff6f45588393d3cbd3c63a79bad97b021b380354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ff6f45588393d3cbd3c63a79bad97b021b380354"}]},"branch":"refs/heads/master"},"f15462e6fb25c1d99790bb7d51c65c9ac66b8314":{"kind":"REWORK","_number":2,"created":"2020-05-26 10:28:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/730736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/730736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/2"}}},"commit":{"parents":[{"commit":"9e48026b1ba04a7bcd04f1083114b497e9199016","subject":"Merge \"Remove redefined constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e48026b1ba04a7bcd04f1083114b497e9199016"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 09:03:47.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 10:28:25.000000000","tz":-420},"subject":"WIP: Add `access_as_readonly` sharing action","message":"WIP: Add `access_as_readonly` sharing action\n\nChange-Id: I3f303583e8b7c56a2759453403eb73940e9ae089\nPartial-Bug: 1875516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f15462e6fb25c1d99790bb7d51c65c9ac66b8314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f15462e6fb25c1d99790bb7d51c65c9ac66b8314"}]},"branch":"refs/heads/master"},"1404f73e74e6f9ed69bed418f6ec26258b69d778":{"kind":"REWORK","_number":3,"created":"2020-08-25 20:43:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/730736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/730736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/3"}}},"commit":{"parents":[{"commit":"fe06e46f7ec3b3285c2754b8f00d69657ef32a29","subject":"Merge \"Log time of the apply_func only if it took more than 0.1 second\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe06e46f7ec3b3285c2754b8f00d69657ef32a29"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 09:03:47.000000000","tz":-420},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-25 20:43:11.000000000","tz":-240},"subject":"WIP: Add `access_as_readonly` sharing action","message":"WIP: Add `access_as_readonly` sharing action\n\nChange-Id: I3f303583e8b7c56a2759453403eb73940e9ae089\nPartial-Bug: 1875516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1404f73e74e6f9ed69bed418f6ec26258b69d778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1404f73e74e6f9ed69bed418f6ec26258b69d778"}]},"branch":"refs/heads/master"},"83019b98b6925998404f0bf37177dac3ff0cd3be":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-03 12:56:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/730736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/36/730736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/36/730736/4"}}},"commit":{"parents":[{"commit":"fe06e46f7ec3b3285c2754b8f00d69657ef32a29","subject":"Merge \"Log time of the apply_func only if it took more than 0.1 second\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe06e46f7ec3b3285c2754b8f00d69657ef32a29"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-26 09:03:47.000000000","tz":-420},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 12:55:42.000000000","tz":-240},"subject":"Add `access_as_readonly` sharing action","message":"Add `access_as_readonly` sharing action\n\nChange-Id: I3f303583e8b7c56a2759453403eb73940e9ae089\nPartial-Bug: 1875516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/83019b98b6925998404f0bf37177dac3ff0cd3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/83019b98b6925998404f0bf37177dac3ff0cd3be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
