)]}'
{"id":"openstack%2Fneutron-lib~730876","triplet_id":"openstack%2Fneutron-lib~master~I7c84a83a2f836493f63b220f7ff0cbcc823374f2","project":"openstack/neutron-lib","branch":"master","topic":"time-cost","hashtags":[],"change_id":"I7c84a83a2f836493f63b220f7ff0cbcc823374f2","subject":"Migrate timecost decorator to neutron-lib","status":"MERGED","created":"2020-05-26 16:30:24.000000000","updated":"2020-06-02 22:27:53.000000000","submitted":"2020-06-02 22:26:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730876-1591136790841-c1c8035f","meta_rev_id":"4e4148da3c6cdfb1177736daba169d1fddbef842","_number":730876,"virtual_id_number":730876,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-05-27 19:09:54.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 22:26:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-06-02 20:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-06-01 12:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-02 20:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 17:25:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-01 12:32:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-06-02 20:41:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-02 22:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c79ac7ccc18b32fad0c424f7c36bddda88a9254","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-26 16:30:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cbafccd9c8da1493a2c163ef9ba06d071643469","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-26 16:40:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8cd76040c4308e4a0cc0493e091a56e86da2082c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 17:25:26.000000000","message":"Patch Set 2: Code-Review+2\n\nMaybe a reno could be needed.","accounts_in_message":[],"_revision_number":2},{"id":"2a7c792c1b11d5b553ffa0b1f77f9e85af1208d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:12:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f386385c96413aa7a340d3bbd95eec : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ba3a162c71042419d681f98023f1842 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e1a8180bce45758f8f0d50784307d4 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a4162a0bf64ba6bdb8f922d09396aa : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09f49015d31481281d1e37b65fe58de : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a014cc7109094ab1957b08789fc2fbf0 : SUCCESS in 5m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b04bb1c1ddd40a1b4a71d99b4d7fe43 : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"98ed3ba623761e27cf76246cfe9c9fbd60ec0578","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-27 09:22:05.000000000","message":"Patch Set 2: Code-Review-1\n\nYes, I agree with Rodolfo, release note would be good to have in this patch. Can You add some short note maybe?","accounts_in_message":[],"_revision_number":2},{"id":"ade0485b1ed2e067d0b2968519bd77d0878d5c0d","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-27 19:09:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f271a03f25c574a65e16edf6ebab66565361016","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-27 19:09:54.000000000","message":"Patch Set 3:\n\n\u003e Yes, I agree with Rodolfo, release note would be good to have in\n \u003e this patch. Can You add some short note maybe?\n\nDone!","accounts_in_message":[],"_revision_number":3},{"id":"b9e542c2ac6faff5c3132c7ee7dd1d8d80e76c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:47:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7440bf11e48140d78a7d2187d78c3a3f : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776e7f16befc48a1a70ab97cc91f205e : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bf2af16460149ae88c2b2ecf649ae6f : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb09d8e46d62431ebdd860f3b7cf8a18 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8c7aff4fe44a6b800855ce0cae0e95 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7e80b7a3954e6cbe368b8b34819a2e : SUCCESS in 7m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38a90b349e424a6cb3c28832e0b7fb7d : SUCCESS in 1h 29m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aac64ff3641438b886b22027fd5159c : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"185ee03a3df7df796a5d4b6d424954002271b9ef","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-01 12:32:23.000000000","message":"Patch Set 3: Code-Review+2\n\nThis way the usage of the decorator is simpler outside of neutron as well","accounts_in_message":[],"_revision_number":3},{"id":"9f42b118d3e180a4e8ca029f0741b5457b32eef3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-02 20:41:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1534039cf6e2359da6a8290dd452740426c28c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 20:41:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10b7858419ebeccd0b9d5d03d7a9b831814fdf8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:26:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b2ada3630245b2b123d7f5c972f926 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cf5aa5387624de38b24fe01c0c83766 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780eaf515b45460a8cbefed0750d7f83 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24bae7cd4dd3442fb342e18ef5589f65 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e2f1bd1cb5f4cdab1919c9268807bfc : SUCCESS in 4m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51a58c08e39f4913bba3f803ab73bcae : SUCCESS in 1h 39m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77b74fd1a17b445b804317ee7eab31a4 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ac66694c78a0ac8b5aee1e1703f5d2675f6c3546","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:26:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"aeb4057dcea4648da434ab37d67ebcce36df6123","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:27:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d696fdedb1f4338aa44e6c651df3383 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afbfe30dfa464ceebabe3a077c671df9 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"492fd109ed92f15e421ebbdbd4fba5a9129eef17","revisions":{"69c53de61abc2cef2b94397aac37825b5beb7e31":{"kind":"REWORK","_number":1,"created":"2020-05-26 16:30:24.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/76/730876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/730876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/1"}}},"commit":{"parents":[{"commit":"9e48026b1ba04a7bcd04f1083114b497e9199016","subject":"Merge \"Remove redefined constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e48026b1ba04a7bcd04f1083114b497e9199016"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:29:18.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:30:23.000000000","tz":-240},"subject":"Migrate timecost decorator to neutron-lib","message":"Migrate timecost decorator to neutron-lib\n\nThere are a few functions in neutron-lib where timecost profiling would\nbe beneficial, so that means it is a good time to transition this\nlibrary to neutron-lib.\n\nChange-Id: I7c84a83a2f836493f63b220f7ff0cbcc823374f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69c53de61abc2cef2b94397aac37825b5beb7e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69c53de61abc2cef2b94397aac37825b5beb7e31"}]},"branch":"refs/heads/master"},"48aeb6b39e9fb561fc41776ac30e156ddf23394b":{"kind":"NO_CHANGE","_number":2,"created":"2020-05-26 16:40:36.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/76/730876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/730876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/2"}}},"commit":{"parents":[{"commit":"9e48026b1ba04a7bcd04f1083114b497e9199016","subject":"Merge \"Remove redefined constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e48026b1ba04a7bcd04f1083114b497e9199016"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:29:18.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:37:30.000000000","tz":-240},"subject":"Migrate timecost decorator to neutron-lib","message":"Migrate timecost decorator to neutron-lib\n\nThere are a few functions in neutron-lib where timecost profiling would\nbe beneficial, so that means it is a good time to transition this\nlibrary to neutron-lib.\n\nChange-Id: I7c84a83a2f836493f63b220f7ff0cbcc823374f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48aeb6b39e9fb561fc41776ac30e156ddf23394b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48aeb6b39e9fb561fc41776ac30e156ddf23394b"}]},"branch":"refs/heads/master"},"492fd109ed92f15e421ebbdbd4fba5a9129eef17":{"kind":"REWORK","_number":3,"created":"2020-05-27 19:09:22.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/76/730876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/76/730876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/76/730876/3"}}},"commit":{"parents":[{"commit":"9e48026b1ba04a7bcd04f1083114b497e9199016","subject":"Merge \"Remove redefined constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e48026b1ba04a7bcd04f1083114b497e9199016"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:29:18.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-27 19:09:09.000000000","tz":-240},"subject":"Migrate timecost decorator to neutron-lib","message":"Migrate timecost decorator to neutron-lib\n\nThere are a few functions in neutron-lib where timecost profiling would\nbe beneficial, so that means it is a good time to transition this\nlibrary to neutron-lib.\n\nChange-Id: I7c84a83a2f836493f63b220f7ff0cbcc823374f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/492fd109ed92f15e421ebbdbd4fba5a9129eef17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/492fd109ed92f15e421ebbdbd4fba5a9129eef17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
