)]}'
{"id":"openstack%2Fneutron-lib~730879","triplet_id":"openstack%2Fneutron-lib~master~I4c8867bb21d01c7a70c1b015814ce5da62e23a50","project":"openstack/neutron-lib","branch":"master","topic":"timecost-resource-helpers","hashtags":[],"change_id":"I4c8867bb21d01c7a70c1b015814ce5da62e23a50","subject":"Check time to apply resource helper functions","status":"MERGED","created":"2020-05-26 16:40:36.000000000","updated":"2020-06-04 18:36:40.000000000","submitted":"2020-06-04 18:32:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730879-1591295563851-ce7c6379","meta_rev_id":"343f35a60e9d3cbd302a065eb16673d8b492667c","_number":730879,"virtual_id_number":730879,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 18:32:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-06-04 14:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-06-04 14:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 14:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-06-04 17:09:09.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 17:32:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-27 06:04:08.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-05-27 09:23:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-04 18:32:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b42f16fd1ba08322128b8a91054d553af86a7c5","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-26 16:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ee1291f9d03a12596c28e57a9fcc35c6ef67e0a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 17:32:11.000000000","message":"Patch Set 1: Code-Review+2\n\nI think most of the extensions will take very short time, but this is not the first time we see an extension killing the performance [1][2] and could help with 3rd party extensions, not tested by default in the CI.\n\n[1]https://review.opendev.org/#/c/667998/6/neutron/services/qos/qos_plugin.py\n[2]https://bugs.launchpad.net/neutron/+bug/1834484","accounts_in_message":[],"_revision_number":1},{"id":"10e792863dc55e70350e80a083a3c1b8d4d24951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:15:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0db94421ab549859352f67c6fef9db9 : SUCCESS in 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31b87d44d692430596e4a31b5b735745 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d55664022ae44d4912844739143083e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1636516d5845af951b48737aef28af : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/891c6eef0aaa4d609f10bfb8a98ccbc5 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/675ac64fc55b416daf275f19d236a212 : SUCCESS in 7m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce1511bb54d443cd8b18c527671b401c : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":1},{"id":"22a0bd629f8e4d56202bb961478debdecf8d040b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-05-27 06:04:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aac3ec6a752ea618dbcbdc0ca7055538f51f0f51","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-27 09:23:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b425a185c011fcf41b2bd73c1f8a2abe665f0a4b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-06-03 21:12:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2dd1388a6745fa190c9c6d0e8ef21a2d6fb7f7ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 22:36:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1d54cde98e45b88cc6ad2822a4a10a : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/059c14236a4342b88498433d283f0ac6 : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b8457d6e7bd4bd48929047fed733404 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6a07a078c741fa88775f39897bdd5d : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4680afff3226413ba4d5e4be9eefb123 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/218bf2b95a6b4b5cbe002e8ed69100d8 : SUCCESS in 4m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e4dcaff1446488abdcd722879228829 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"234d1e68ff1e8ed66b6ff63a3cd47dfbf43b99a3","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-06-04 14:55:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2b4fc6db2ebeb9b281c1482a1364c89339b47600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:51:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6317f4659c224c79907b7765b7e024a3 : SUCCESS in 8m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65f4de28aff14c52a396067dec69baec : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23a05d5673414d9cbefd1d8e4a095645 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d310633f097b43479f9dbce3bb62acb0 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc133652cf1541f09545039ae6f44335 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec4b08335d3422f8345fdeeafc017de : SUCCESS in 5m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f491a878d5bd4e74ae861c0025674468 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f19e2c951c0974f4fe263c40f2257e5216967b31","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-06-04 17:09:09.000000000","message":"Patch Set 2: Workflow+1\n\nReinstating previous +W from Slawek","accounts_in_message":[],"_revision_number":2},{"id":"8ba63f2bdcba1c94cfe23bf0b401908790d2871d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:09:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc9d2d9ed515abedda6e2cd357b476c5169abf74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:32:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a37aef388474d23b78c19c87cc2e5a0 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a96d0e0df4447283502632b73e3b97 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9dfff61a40149e0bb33c66e6430b3ce : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512be414736e40cb95f0c2ce50a597f7 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a8544e01b794d08acef7c0b2c315a63 : SUCCESS in 6m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3824a8c0fe64c31902655cbccc5ac1d : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8d2ae24106414f925787307a06fd127e554dad9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:32:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e7358e2788289a0accc0469c56eb1f6d83f85436","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:36:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c44ef7fceb4253a390db697ea43d14 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29ba0b7d2172ce890a4113105ca931b72d6e3f59","revisions":{"a302ff0c14fd0a494d2cf65fe0f8a79e5edf7e8c":{"kind":"REWORK","_number":1,"created":"2020-05-26 16:40:36.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/79/730879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/730879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/1"}}},"commit":{"parents":[{"commit":"48aeb6b39e9fb561fc41776ac30e156ddf23394b","subject":"Migrate timecost decorator to neutron-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48aeb6b39e9fb561fc41776ac30e156ddf23394b"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:37:56.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:40:26.000000000","tz":-240},"subject":"Check time to apply resource helper functions","message":"Check time to apply resource helper functions\n\nFor many operations, the time it takes to apply resource helper\nfunctions constitutes a significant percentage of the time spent\ncompleting the total transaction.  A list port transaction may have to\nprocess thirteen resource helpers, for example.  Introducing some timing\nlogging will allow us to get clarity on whether slowdowns that happen\nfrom version to version are a result of these accumulated resource\nextenders.\n\nChange-Id: I4c8867bb21d01c7a70c1b015814ce5da62e23a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a302ff0c14fd0a494d2cf65fe0f8a79e5edf7e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a302ff0c14fd0a494d2cf65fe0f8a79e5edf7e8c"}]},"branch":"refs/heads/master"},"29ba0b7d2172ce890a4113105ca931b72d6e3f59":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-04 14:55:27.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/79/730879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/730879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/730879/2"}}},"commit":{"parents":[{"commit":"b91f9fc0cf6908a28a73500e58d0505fcd46983e","subject":"Merge \"Migrate timecost decorator to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b91f9fc0cf6908a28a73500e58d0505fcd46983e"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-26 16:37:56.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-06-04 14:55:27.000000000","tz":0},"subject":"Check time to apply resource helper functions","message":"Check time to apply resource helper functions\n\nFor many operations, the time it takes to apply resource helper\nfunctions constitutes a significant percentage of the time spent\ncompleting the total transaction.  A list port transaction may have to\nprocess thirteen resource helpers, for example.  Introducing some timing\nlogging will allow us to get clarity on whether slowdowns that happen\nfrom version to version are a result of these accumulated resource\nextenders.\n\nChange-Id: I4c8867bb21d01c7a70c1b015814ce5da62e23a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/29ba0b7d2172ce890a4113105ca931b72d6e3f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/29ba0b7d2172ce890a4113105ca931b72d6e3f59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
