)]}'
{"id":"openstack%2Fneutron-lib~732141","triplet_id":"openstack%2Fneutron-lib~master~I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af","subject":"[OVN] Add distributed device_owner","status":"MERGED","created":"2020-06-01 10:04:11.000000000","updated":"2020-06-16 21:45:30.000000000","submitted":"2020-06-16 21:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732141-1592343816956-d12a0952","meta_rev_id":"bcc45bc3ad7282c9d5450c5d528b2f4c141a12b1","_number":732141,"virtual_id_number":732141,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-06-16 15:01:31.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 21:43:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-06-16 19:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-06-16 15:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-06-16 19:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 07:46:04.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-06-08 08:40:51.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-06-08 08:41:08.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-06-16 08:37:10.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-06-16 08:53:04.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-06-16 09:09:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-16 12:40:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-06-16 14:15:32.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-16 15:22:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-16 19:57:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-06-16 21:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b19f5320eb881d9b30d4ddbd4a556fa5993a0763","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-01 10:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef95a4620cb79604d6c1f2979ec39eada7671a1","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-01 10:12:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nJust some nits on the commit message itself and a conceptual reflection :)\nLet me know what you think, please.","accounts_in_message":[],"_revision_number":1},{"id":"46cc83dd3a16bce96ce0cc4e3b6bcd8aaa79d9c8","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-01 10:18:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85af0ed90e40ef0a89476f1577378c8d3f585539","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-01 10:21:44.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks for fast review Daniel!","accounts_in_message":[],"_revision_number":1},{"id":"e994b35bce0519ea4239b624d70879d7750f455b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-01 10:39:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"786b56cdbf93ec4cd2cfce159fe36d3ca809a2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 12:07:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f2e069677c4f649955c9eb972b7aac : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ffa710d8869417e82496b742f4c8a5b : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7d6a43c76f4d75a77e0da775a85b0e : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131279d43f3c449b8aa4b014ae27f7bc : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ce3ee753b8443fa2675022bbaa92b5 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f1bf90a76144bb593058b76eef705c2 : SUCCESS in 4m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50a8ef0407d94c01932f4c9d920f142f : SUCCESS in 1h 26m 53s","accounts_in_message":[],"_revision_number":2},{"id":"52e1e1d186a8f47df28af761f350ee9d1ec2f1bb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-01 14:46:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7b8affd7eea6c2b46afb96ab54438eda3fab737a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-03 09:40:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5f71cfad3c816af25bdd6dc0f47e902f8e7372b8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 21:21:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"159d707cdf55f7e65f4966163b86aeacbacf2d4f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-08 07:45:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0f77879dca11bdc16d64889033b751cf100deea","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-08 08:41:43.000000000","message":"Patch Set 2:\n\nHello PTG Attendees!\n\nPlease take a look on this patch after our discussion around having \u0027distributed\u0027 device_owner for distributed services, like OVN Metadata.","accounts_in_message":[],"_revision_number":2},{"id":"b6b284ec23af277f4dfcc455a7ae373590e49ac5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-15 08:31:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2871cb8799635b7743904bfc35b8dd5c8ba5aac6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 08:01:13.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b7b34ff46f3b9a4d6bbd5a78e1247eed16bafcb5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-16 08:37:10.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Maciej! neutron:distributed makes sense to me","accounts_in_message":[],"_revision_number":3},{"id":"df5a05e25e7256624391edee13b1945bc968e775","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-16 08:48:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22a681397c12e4f657d7f8a4282b380c2c48100e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 08:50:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"767d07c82f17efe83013b06796e3779ae96722d0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 08:50:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bbd582c31101c54eeb4d13bf8eb50716e07d95bb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-06-16 08:53:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0b02f645b235f372dd11b1b19a92a3e605e831b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 09:09:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ddd2609975d986c2b5b2f83bd4cff63dad9264db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:27:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0982e58281a947fc9bae64eba9acb7a3 : SUCCESS in 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/310800ed02384dce888b19095ecebde0 : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ef89df5f01e42e0885f8e55fb7b61f9 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14376c8728b947ccaba566f8461ca93d : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/286147281eec4df19fc81c3b76720a9d : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24d5a81e94d94430b10322d0dcdc93c5 : SUCCESS in 8m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c917db9171924de3a202dcfff18273f1 : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e176ac01777669cb2099281a551f58b02b3275b0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-16 12:40:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bcd19176a7e6355004b4adb7f3614ac2719b2169","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-16 12:41:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1771a9291c8878843fad4a4b10a2c4cc366590e0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-16 14:15:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f64f37670e4fc60b7bb2d173c7e34c2d6271e880","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 15:00:41.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f1d1448890335ff4d177931c668164f2fb414e1f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 15:01:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7fcef855e3ae1a27bd7d3d33f522a7c8bf1267ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-16 15:22:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bac90c661417f2c6217173fed88b4bdd822d5c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 16:19:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7796f965876e4fb6bb2ea2fc435750fc : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6bc65244bd14dc2857a67ebe493d14e : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01bf5f0d1ad048fb9a74e82657c79640 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/367ffa55beee448bb56bb2352ab8dae2 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59baed32a2b84da3a580729223e4deeb : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0296b92b02b4fcf80d4f72aa175dfd2 : SUCCESS in 8m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e2b3938a4d041768e40279f9b5a4552 : SUCCESS in 1h 12m 50s","accounts_in_message":[],"_revision_number":5},{"id":"8750ef3f4ae81525375c5b2449e79161603e0dba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 19:57:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6f20a26f19b783b0426ffd46c2532a64dd1ab1f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 19:57:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"375d307341c7577cda7ea27e712faabe8f3f5026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:43:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"582f89605df40a6c879b2012c601d6bfc2bf3a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:43:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93665adf3cec44f3a376f6f876a5c891 : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f71035ee8a57473a8c0da7d1928d73d1 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2683831d5f8446d99eb8d8490aba921 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cee9b1b354704d2c967c02f5b843eb78 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/538a3352c2ba431b88cb5d308b833162 : SUCCESS in 7m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3c107e2c5e649048abb3547b5703cae : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":5},{"id":"bcc45bc3ad7282c9d5450c5d528b2f4c141a12b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:45:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e715ae8ffe2a4ea987e54b80b256c8ab : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"438a9b7609f7367f33a5d61c6111e5b81ce06cf6","revisions":{"25e23e2e47f334e8b322b9e3bcb2b93e87325dba":{"kind":"REWORK","_number":1,"created":"2020-06-01 10:04:11.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/41/732141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/732141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/1"}}},"commit":{"parents":[{"commit":"e1b0ae910b9aaf97576da9f17c5fded6e7872eff","subject":"Merge \"Rehome abstract class LinuxInterfaceDriver definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1b0ae910b9aaf97576da9f17c5fded6e7872eff"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 09:48:09.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 10:02:33.000000000","tz":0},"subject":"[OVN] Add distributed device_owner","message":"[OVN] Add distributed device_owner\n\nAdd distributed device owner for services like OVN\nMetadata and OVN DHCP.\n\nOVN Metadata uses \u0027localport\u0027 port concept in OVN [1]:\n\n\"A connection to a local VIF. Traffic that  arrives  on  a\nlocalport  is  never  forwarded  over a tunnel to another\nchassis. These ports are present  on  every  chassis  and\nhave  the  same  address  in all of them. This is used to\nmodel connectivity to local services that  run  on  every\nhypervisor.\" [1]\n\nFor OVN DHCP we use localport fixed-ip allocation to get\nIP address used as `server_id`, but the DHCP port is not bound\nanywhere.\n\nBecause of those two reasons we should define new device_owner\nand threat it differently in Neutron API, to unlock the OVN\ndistributed services potential and to not use the same port_type\nfor Neutron not-distributed services.\n\n[1] https://www.ovn.org/support/dist-docs/ovn-nb.5.txt\n\nChange-Id: I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/25e23e2e47f334e8b322b9e3bcb2b93e87325dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/25e23e2e47f334e8b322b9e3bcb2b93e87325dba"}]},"branch":"refs/heads/master"},"1c674c01a88999a0003e5e9b265bc641a3b147f3":{"kind":"REWORK","_number":2,"created":"2020-06-01 10:39:50.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/41/732141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/732141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/2"}}},"commit":{"parents":[{"commit":"e1b0ae910b9aaf97576da9f17c5fded6e7872eff","subject":"Merge \"Rehome abstract class LinuxInterfaceDriver definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1b0ae910b9aaf97576da9f17c5fded6e7872eff"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 09:48:09.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 10:36:43.000000000","tz":0},"subject":"[OVN] Add distributed device_owner","message":"[OVN] Add distributed device_owner\n\nAdd distributed device owner for services like OVN\nMetadata and OVN DHCP.\n\nOVN Metadata uses \u0027localport\u0027 port concept in OVN [1]:\n\n\"A connection to a local VIF. Traffic that  arrives  on  a\nlocalport  is  never  forwarded  over a tunnel to another\nchassis. These ports are present  on  every  chassis  and\nhave  the  same  address  in all of them. This is used to\nmodel connectivity to local services that  run  on  every\nhypervisor.\" [1]\n\nFor OVN DHCP we use localport fixed-ip allocation to get\nIP address used as `server_id`, but the DHCP port is not bound\nanywhere.\n\nBecause of those two reasons we should define new device_owner\nand treat it differently in Neutron API, to unlock the OVN\ndistributed services benefits and to not use the same port_type\nfor Neutron non-distributed services.\n\n[1] https://www.ovn.org/support/dist-docs/ovn-nb.5.txt\n\nChange-Id: I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c674c01a88999a0003e5e9b265bc641a3b147f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c674c01a88999a0003e5e9b265bc641a3b147f3"}]},"branch":"refs/heads/master"},"0fbed1d6839dde5bd763e61424d8f15eda713501":{"kind":"REWORK","_number":3,"created":"2020-06-16 08:01:13.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/41/732141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/732141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/3"}}},"commit":{"parents":[{"commit":"e1b0ae910b9aaf97576da9f17c5fded6e7872eff","subject":"Merge \"Rehome abstract class LinuxInterfaceDriver definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1b0ae910b9aaf97576da9f17c5fded6e7872eff"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 09:48:09.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-16 08:01:10.000000000","tz":0},"subject":"[OVN] Add distributed device_owner","message":"[OVN] Add distributed device_owner\n\nAdd distributed device owner for services like OVN\nMetadata and OVN DHCP.\n\nOVN Metadata uses \u0027localport\u0027 port concept in OVN [1]:\n\n\"A connection to a local VIF. Traffic that  arrives  on  a\nlocalport  is  never  forwarded  over a tunnel to another\nchassis. These ports are present  on  every  chassis  and\nhave  the  same  address  in all of them. This is used to\nmodel connectivity to local services that  run  on  every\nhypervisor.\" [1]\n\nFor OVN DHCP we use localport fixed-ip allocation to get\nIP address used as `server_id`, but the DHCP port is not bound\nanywhere.\n\nBecause of those two reasons we should define new device_owner\nand treat it differently in Neutron API, to unlock the OVN\ndistributed services benefits and to not use the same port_type\nfor Neutron non-distributed services.\n\n[1] https://www.ovn.org/support/dist-docs/ovn-nb.5.txt\n\nChange-Id: I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0fbed1d6839dde5bd763e61424d8f15eda713501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0fbed1d6839dde5bd763e61424d8f15eda713501"}]},"branch":"refs/heads/master"},"a14e812be5e21e4234c147a19046363e043c343a":{"kind":"REWORK","_number":4,"created":"2020-06-16 08:50:23.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/41/732141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/732141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/4"}}},"commit":{"parents":[{"commit":"e1b0ae910b9aaf97576da9f17c5fded6e7872eff","subject":"Merge \"Rehome abstract class LinuxInterfaceDriver definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1b0ae910b9aaf97576da9f17c5fded6e7872eff"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 09:48:09.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-16 08:50:20.000000000","tz":0},"subject":"[OVN] Add distributed device_owner","message":"[OVN] Add distributed device_owner\n\nAdd distributed device owner for services like OVN\nMetadata and OVN DHCP.\n\nOVN Metadata uses \u0027localport\u0027 port concept in OVN [1]:\n\n\"A connection to a local VIF. Traffic that  arrives  on  a\nlocalport  is  never  forwarded  over a tunnel to another\nchassis. These ports are present  on  every  chassis  and\nhave  the  same  address  in all of them. This is used to\nmodel connectivity to local services that  run  on  every\nhypervisor.\" [1]\n\nFor OVN DHCP we use localport fixed-ip allocation to get\nIP address used as `server_id`, but the DHCP port is not bound\nanywhere.\n\nBecause of those two reasons we should define new device_owner\nand treat it differently in Neutron API, to unlock the OVN\ndistributed services benefits and to not use the same port_type\nfor Neutron non-distributed services.\n\n[1] https://www.ovn.org/support/dist-docs/ovn-nb.5.txt\n\nChange-Id: I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a14e812be5e21e4234c147a19046363e043c343a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a14e812be5e21e4234c147a19046363e043c343a"}]},"branch":"refs/heads/master"},"438a9b7609f7367f33a5d61c6111e5b81ce06cf6":{"kind":"REWORK","_number":5,"created":"2020-06-16 15:00:41.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/41/732141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/732141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/732141/5"}}},"commit":{"parents":[{"commit":"e1b0ae910b9aaf97576da9f17c5fded6e7872eff","subject":"Merge \"Rehome abstract class LinuxInterfaceDriver definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e1b0ae910b9aaf97576da9f17c5fded6e7872eff"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-01 09:48:09.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-16 15:00:30.000000000","tz":0},"subject":"[OVN] Add distributed device_owner","message":"[OVN] Add distributed device_owner\n\nAdd distributed device owner for services like OVN\nMetadata and OVN DHCP.\n\nOVN Metadata uses \u0027localport\u0027 port concept in OVN [1]:\n\n\"A connection to a local VIF. Traffic that  arrives  on  a\nlocalport  is  never  forwarded  over a tunnel to another\nchassis. These ports are present  on  every  chassis  and\nhave  the  same  address  in all of them. This is used to\nmodel connectivity to local services that  run  on  every\nhypervisor.\" [1]\n\nFor OVN DHCP we use localport fixed-ip allocation to get\nIP address used as `server_id`, but the DHCP port is not bound\nanywhere.\n\nBecause of those two reasons we should define new device_owner\nand treat it differently in Neutron API, to unlock the OVN\ndistributed services benefits and to not use the same port_type\nfor Neutron non-distributed services.\n\n[1] https://www.ovn.org/support/dist-docs/ovn-nb.5.txt\n\nChange-Id: I53828ec6e1f0d2861eebfadcb9d46551fbf1a6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/438a9b7609f7367f33a5d61c6111e5b81ce06cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/438a9b7609f7367f33a5d61c6111e5b81ce06cf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
