)]}'
{"id":"openstack%2Fneutron-lib~738205","triplet_id":"openstack%2Fneutron-lib~master~Ic382feb7fca036b3c7aae5f21399b7639520a92b","project":"openstack/neutron-lib","branch":"master","topic":"metadata-ipv6","hashtags":[],"change_id":"Ic382feb7fca036b3c7aae5f21399b7639520a92b","subject":"Metadata addresses","status":"MERGED","created":"2020-06-26 13:34:01.000000000","updated":"2020-07-13 19:48:12.000000000","submitted":"2020-07-13 19:46:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738205-1594669580602-eb87a29e","meta_rev_id":"f63cc9ce3e050639a706017f0cf6b2fc91055f8d","_number":738205,"virtual_id_number":738205,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 19:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2020-07-10 12:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-07-10 11:43:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-07-13 18:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-07-13 18:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 16:20:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-07-10 11:43:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-07-10 12:55:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-13 18:26:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-13 19:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e36951444114c58dffd6d7d380a22d19e543c24c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-06-26 13:34:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6bd4404096bddacabafdf5caae64984934fd06b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-06-26 13:35:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0596842d2ed3d940b80338fc5a1ec172a3bbc00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6525f9925f44274a6e1863e9782b3f6 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9d22a27868343e6a4f847de3b871806 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43da3e7967be421ebea56b0402c0ad33 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a6f709df97a4ed0b4a2dfcd73258f18 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9fc6022849a4609b71edbf49cc67fb2 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/023556a013544a2ea88324a6e301a360 : SUCCESS in 4m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd086abeb7bf4862a42c0bf77f8e0385 : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"28c86732bb0e39d36a822b11d19a54bdb2627c6c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-26 17:37:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\n-1 just because I want to make sure we\u0027re using the different formats of the metadata address correctly.  I posted https://review.opendev.org/738247 to help make things better, and possibly fix some mistakes.","accounts_in_message":[],"_revision_number":1},{"id":"cc2838acbbf49450b312316c971ed82abc6aef37","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-07 08:48:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9a02eabe5b0ae274730b6523d24b86c21d443bb4","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-07-07 13:09:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcd8c7899b1401b6329740ff677982fc2907204b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-07-07 13:12:52.000000000","message":"Patch Set 2:\n\nLet me know please if I missed something.","accounts_in_message":[],"_revision_number":2},{"id":"202ec0c6b5f7eab8736cc6b4d144298981c2240c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 14:07:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc042fbb0a0d4a3d9d88cf1a3747ce1d : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0f0d3cd5a046bb87de75569260de5f : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10876ee70ebb45dca68f20beea15c91c : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71b606711e94b678f82c2b371bd67cb : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a4f649d7015443c9a94b31ae52a930b : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e14b309a43c48bbb14d55d97168c635 : SUCCESS in 7m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b7794a17ccd4287a58f36554e983f28 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8f11ca7ab2ffeb7edf2e90c63da9b74b039d1e7b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-07 16:20:30.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"afdcec6e92dd72e8b7d6a9f9cd92106807a9eef9","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-07-09 08:12:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f48e6616f2d4340b38bef6c242426fd61d5321d6","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-07-09 08:12:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fc2009ba57be58d6ffa222ce4b7f04ddb684db7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:53:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23147464d95846b3ac847d2577460718 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2df0c25ad6d4e79a95b2430d56ec13f : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e5c99c9ce740b5a1c042a3f70fc538 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581a0f9abf3540e5845e0c52b84260de : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ef144f5a01142339f85a60f291a7d56 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1ba236282e42239843752cc90721d4 : SUCCESS in 7m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa4695cfeca942869bb57d639438bed3 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3104bd0761450167645d1a31fad338f0fd85216c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-10 11:43:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8698c50c8fba10f8942aeb7a56ef742e4d64cd8e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-10 12:55:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5255b4a1a3dd74f9103fa8e34a634189b77c3def","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-13 18:26:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f16c32fa810511367a3bba82d7b33a1852d58cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 18:26:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1636e583d25404604ed88f55f16e477117f316c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:46:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f199d09b024a10e49d24857139cac324b9eeac3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:46:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49dd16f407914df6827b65d0459b16a1 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f5a9d2e17704500a2232f63baaaeab9 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc1f09b4189493780655afa6126057c : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e490ea45ddce459f8efe29cbe12cebba : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/752ad15161df4c8d9782c8ad29118166 : SUCCESS in 4m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75fcba32e2ae4a65adc4cdcae9bb7c02 : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a02c7f5691bb8cc8b0d4d0b00c42e60251dd6cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:48:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49772c4e5882450da9ce8861b15e4c7b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"60eab578fc74bf1350539f7d4729fb05abf5e0c8","revisions":{"2ab4f429ac357255af5788a39284d5c7d246ab9c":{"kind":"REWORK","_number":1,"created":"2020-06-26 13:34:01.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/05/738205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/05/738205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/1"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-06-26 13:30:06.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-06-26 13:30:06.000000000","tz":120},"subject":"Metadata addresses","message":"Metadata addresses\n\nDefine well-known metadata addresses in neutron-lib so they are easily\nshared between various implementations like dhcp-agent and ovn.\n\nChange-Id: Ic382feb7fca036b3c7aae5f21399b7639520a92b\nPartial-Bug: #1460177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2ab4f429ac357255af5788a39284d5c7d246ab9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2ab4f429ac357255af5788a39284d5c7d246ab9c"}]},"branch":"refs/heads/master"},"47db0366a2fff0729c299d3c85abeda84d2437be":{"kind":"REWORK","_number":2,"created":"2020-07-07 13:09:15.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/05/738205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/05/738205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/2"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-06-26 13:30:06.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-07-07 13:06:23.000000000","tz":120},"subject":"Metadata addresses","message":"Metadata addresses\n\nDefine well-known metadata addresses in neutron-lib so they are easily\nshared between various implementations like dhcp-agent and ovn.\n\nChange-Id: Ic382feb7fca036b3c7aae5f21399b7639520a92b\nPartial-Bug: #1460177\nRelated-Change: https://review.opendev.org/738247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/47db0366a2fff0729c299d3c85abeda84d2437be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/47db0366a2fff0729c299d3c85abeda84d2437be"}]},"branch":"refs/heads/master"},"60eab578fc74bf1350539f7d4729fb05abf5e0c8":{"kind":"REWORK","_number":3,"created":"2020-07-09 08:12:24.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/05/738205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/05/738205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/05/738205/3"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-06-26 13:30:06.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2020-07-09 08:08:51.000000000","tz":120},"subject":"Metadata addresses","message":"Metadata addresses\n\nDefine well-known metadata addresses in neutron-lib so they are easily\nshared between various implementations like dhcp-agent and ovn.\n\nChange-Id: Ic382feb7fca036b3c7aae5f21399b7639520a92b\nPartial-Bug: #1460177\nRelated-Change: https://review.opendev.org/738247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60eab578fc74bf1350539f7d4729fb05abf5e0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60eab578fc74bf1350539f7d4729fb05abf5e0c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
