)]}'
{"id":"openstack%2Fneutron-lib~738541","triplet_id":"openstack%2Fneutron-lib~master~Ifecbaf54a0aafe8bdc1cb0c4b373bef9f5450e27","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"Ifecbaf54a0aafe8bdc1cb0c4b373bef9f5450e27","subject":"Remove fwass_v2","status":"ABANDONED","created":"2020-06-30 02:51:10.000000000","updated":"2020-09-25 15:28:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f996c2fef7b846ccbd529e020903b7f2928b7a6e","_number":738541,"virtual_id_number":738541,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-25 15:28:38.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-30 06:37:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2020-06-30 15:03:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 06:37:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 15:03:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-09-25 15:28:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5c2c4d41cf1f42954428b7a26ee749351d43f488","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-06-30 02:51:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6a136810bd3afa61d6c18aea31fecc71083c067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 04:03:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15884c4071bb434e830935a6c1659238 : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64ec78658cd44eeba2f8e0c597153b7a : FAILURE in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/880b8f2334fb457785c84dce01c8b658 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e80b98e5c04c8e9278834632f07648 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d8fdbc6090d46b2b359d8bd1e122efa : FAILURE in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31469b85017a4c2cb837a1c86965a1d0 : FAILURE in 5m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e72d1818c4124bdaaa508d3faa29a3ac : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0f4c532cf2b5eab5ae1756bbda9ff837ef024965","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-06-30 05:40:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be1ee13c2a198be67396c7c05096fbbce49a9520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 06:37:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7383c4dec54c45bd19f066270f1fbc : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e242d7096ea5491a8dfe07c9ba5003b5 : FAILURE in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb65801b864e4a24a6970adbbb3722eb : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9acc2bb45f254fb4bde38a60787ab152 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb44b1904b8431f8bd3b7f478a240b6 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6062047ceca34c749d642f493d1beeea : FAILURE in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4600e6831544950be8a7a5cd5ffca2e : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8d7abf876eb9ef05194a4492ce1b37f9b70d02c8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-30 15:03:58.000000000","message":"Patch Set 2: Code-Review-1\n\nbased on 19.06.2020 drivers meeting the api-defs should left here, see the logs:\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-meeting/%23openstack-meeting.2020-06-19.log.html#t2020-06-19T14:30:01","accounts_in_message":[],"_revision_number":2},{"id":"f996c2fef7b846ccbd529e020903b7f2928b7a6e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:28:38.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d27a07ae7d34d73d3ab8da138ddc492eccd4102","revisions":{"71d9dfbca1b0671e09aaeaa558cd71874e7d83d2":{"kind":"REWORK","_number":1,"created":"2020-06-30 02:51:10.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/41/738541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/738541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/1"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-06-24 23:46:43.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-06-25 00:05:30.000000000","tz":-240},"subject":"Remove fwass_v2","message":"Remove fwass_v2\n\nSince neutron-fwaas has been retired on the master branch[1], it\n(related API definitions, exceptions, etc.) can be removed.\n\n[1] https://review.opendev.org/#/c/735829/\n\nChange-Id: Ifecbaf54a0aafe8bdc1cb0c4b373bef9f5450e27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/71d9dfbca1b0671e09aaeaa558cd71874e7d83d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/71d9dfbca1b0671e09aaeaa558cd71874e7d83d2"}]},"branch":"refs/heads/master"},"0d27a07ae7d34d73d3ab8da138ddc492eccd4102":{"kind":"REWORK","_number":2,"created":"2020-06-30 05:40:32.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/41/738541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/738541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/738541/2"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-06-24 23:46:43.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-06-25 02:54:32.000000000","tz":-240},"subject":"Remove fwass_v2","message":"Remove fwass_v2\n\nSince neutron-fwaas has been retired on the master branch[1], it\n(related API definitions, exceptions, etc.) can be removed.\n\n[1] https://review.opendev.org/#/c/735829/\n\nChange-Id: Ifecbaf54a0aafe8bdc1cb0c4b373bef9f5450e27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0d27a07ae7d34d73d3ab8da138ddc492eccd4102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0d27a07ae7d34d73d3ab8da138ddc492eccd4102"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
