)]}'
{"id":"openstack%2Fneutron-lib~741452","triplet_id":"openstack%2Fneutron-lib~master~I9dab2e9e9728183ee209a038a0347eb7b4a83a1c","project":"openstack/neutron-lib","branch":"master","topic":"bug/1882804","hashtags":[],"change_id":"I9dab2e9e9728183ee209a038a0347eb7b4a83a1c","subject":"Add placement client methods for allocations","status":"MERGED","created":"2020-07-16 13:48:41.000000000","updated":"2020-09-04 00:14:24.000000000","submitted":"2020-09-04 00:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741452-1599178382779-9d651cf1","meta_rev_id":"e8ee90b8186a30f91f3b583fdefea3f8ade96ee5","_number":741452,"virtual_id_number":741452,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 00:13:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-09-03 22:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-09-03 20:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-09-03 22:50:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 15:09:19.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-08-31 13:17:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-09-03 16:01:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-03 20:09:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-03 20:13:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-03 22:50:48.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-09-04 00:13:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcd5991e8847d4464e0035d31e99f84e8e06f803","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-16 13:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5f5e0ef60fe662fb6da540c8c601f7e4e71389e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:00:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723c6e984d514a8cae89f8dba9d22bce : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee4414bcb94548fd84706443819e6b6c : FAILURE in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a6b3cc6adb244589a990cfb9a66450c : FAILURE in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a5bf8cc74c496a81b8082b2fe14ddf : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972216c5470449c3b5181162087cc019 : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93e2147aef4a4cb093242aa86fed4a8f : FAILURE in 4m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43c5c73c0e634252a582d62c9d46b57e : FAILURE in 1h 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9425ae944a442b3898a200e5fea77bf : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"07397a444a0c48c6471a7d3ba84cd3544b609f64","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-17 08:49:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bdfa29deb99529f2a34d7f5c603e0621bb0e46f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 10:14:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fbb944db7b430fbd7bd772faf7fa14 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24fcc0bb038c45cb9cd972125f543998 : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1b21d1b06684c089841e87ec46a376a : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82515c7f4644439fa72dfc7ed440ac4d : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aee8d959c7d430aab119200a5ff2ff0 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d63c6c8af1d4c72a4351627e4064e71 : SUCCESS in 6m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5782d5b23b394a908a7665184475f882 : SUCCESS in 1h 20m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f46bcc5c3d254feca559d633804868fc : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0252a2d532a5a8765e445a2fd9234f2b27ad5080","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-17 14:01:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d56b453fdaf183b7d7e2df0997f7b3268bf08d4a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-31 13:37:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9437838782012f8040deda692c04d63880f3b92c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-03 15:12:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"113d9aad6cbf7f3c7cd3d82199a795efcdc1a9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:28:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f64fb3871164d4691907c9987c4f768 : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16114fb012404179b9199cbf42f700f4 : FAILURE in 3m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4e8b52553da47a2b5c5a481d33b3ede : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3423bc1046c24657b4328d9971d803d2 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a435fc10ec448e2bb5d0756e9f84b20 : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/399c906776dd4be59103bd69193d7687 : FAILURE in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aab8c13f143846da96a94f0a3ef5e5ae : SUCCESS in 1h 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4636376d92024326926485cc2c956030 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6e3928722c1e4fe23b39c79809fb06fa97ac3082","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 10:39:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ae85284191ca6d25832aaddfb9e099cf06856e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:20:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57a5391f3f346899c884476c85e8633 : SUCCESS in 12m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5968d88b86b4cb8aaa794c0f5a6985d : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1821edaef9cb4c6cac0ed62a547ab07a : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4296cbcd6d084f5aa71c7dcb684a14db : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4255252805142a4878de9b7d3b7a973 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3de16082ed3e4d83a311e6ca7e1cfa6d : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc407eda5d9a43e78055d1829228998e : SUCCESS in 1h 38m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/073894143b0e49ccaa9e99c760666865 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"dfed300bd15af22777ec74fb65dbacc73abcdace","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-06 08:26:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nThanks Lajos! It is a lot better now. -1 only due to the missing unit test coverage. \n\ncheers,\ngibi","accounts_in_message":[],"_revision_number":4},{"id":"1d6b3b71e77c809624e4c66fb5c8af7e3e06dc88","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-07 09:47:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c2725a78676c4b8fcef33c21abe9a4f2d745c84f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-07 09:49:17.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks for the review","accounts_in_message":[],"_revision_number":5},{"id":"7a54f7b6a48ed717bddbb62dd12a0f1283476411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:16:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a1ea8e4b354d8180ccf6e8ba5e24fb : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/781a0c72ef7a46d495a7df031300d1cd : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57697c7c2b9d4992a7723f7a2e579edb : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46ee4fb98ba49299343d59e58c41a92 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26581d7aca744eff9557eead801f1c75 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ee728f4a9a1475ea5055400ebe6a8fb : SUCCESS in 4m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40f17bc25880448b8c6adfa35a6f2a05 : SUCCESS in 1h 22m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8148927fe3a49899103f6a872534907 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":5},{"id":"6fdea36b2f83908632f8cb507ff033a162a7615b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-07 15:09:19.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nThe placement interaction part looks good to me.","accounts_in_message":[],"_revision_number":5},{"id":"4b4ea341487fabcbe45c175693ea98c233d24335","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-19 11:33:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"61940527e64468cd3b01749704da51122caea3c6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 10:10:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3528b0cc0294eafcc0354d1df775ff508ecec9f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-31 13:17:52.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)\n\nminor nits","accounts_in_message":[],"_revision_number":5},{"id":"023939ee58809c8226b750336f7d07f16f1d7eb5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-03 10:27:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3950595fe1091e1171b3d652d167914978d9aac8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-03 13:48:57.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"81ecd1ccb5c9d882762f494401f6d53f8ccfc1dc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-03 13:50:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"87bb2a8a66690be3bb9444739b5e07ba86e76572","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 14:20:43.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b276f9b5ed7dc32c537a3bb9ab0b1f0e7b3ffc1e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-03 14:23:59.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"22d6665e5e9569dff388a0c6f56e6bc01ac6aaaa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-03 14:24:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"426d15d2748c4e999534aaf177b41e172cabd45a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-03 16:01:22.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nI followed the discussion so far and it looks good.","accounts_in_message":[],"_revision_number":7},{"id":"8c86142847f316f7264ede8aaf5fc653b10ee607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:08:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c828901420704339b0d1d2f9ecbfc0ac : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2991359145e744e499adc8fd6496b3f6 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef70c30116494f1f968097c74f4afd65 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc7721ff65045de9fda332aa22f74e1 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3f3121c7c704b708face55c81f927cf : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ca55d99bc6145cb86f13a24541a75e8 : SUCCESS in 6m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/950427a1b2a64a5e85eb4e6a4a0a7e41 : SUCCESS in 1h 40m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c206c476ab4bd7a675727dcefabf2b : SUCCESS in 4m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c74bacd5ab284fd6ad6c3cf9a8cffb04 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":7},{"id":"97cb50b6edc6db6c43ca256a001bf0445d43ccb4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:13:00.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a3d55b3a5504dbcb23be622b42dc4b76cb98cb16","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-03 20:09:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"27050d45a9d5064bfd623c9c451d8fbb589424d3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-03 20:09:55.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2358137610adf042fa33d2457452864573e941e3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 20:13:02.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c7e858a72c43a9ace82233d7c6ab9c332fe37b88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 20:13:27.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM, just fixed missing space in relnote","accounts_in_message":[],"_revision_number":9},{"id":"7b47f36d883283649ae4453bbb0f00b1fe614bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:40:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e50c31ac564ebe99891e0e82f72ae8 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1968da07604fa9b28dd3ce9ee19e76 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc46174b0e64f438551ec10f8b10a6a : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8e243777054ea9be9db1f39c3d7cba : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5f92b6149af408aa2a002f61c498c58 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9150b978dee4964aa0577b733547dab : SUCCESS in 6m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4df4a33720954202b435b2239f2532b7 : SUCCESS in 1h 18m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cea3d7894ab34fcfa13a6617b9a90aad : SUCCESS in 5m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0091ed55e6974924a1bf3c3fceb64682 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":9},{"id":"7ee4a55214b084c97cf2ba879abd2f64dfc8d5ca","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-09-03 22:50:48.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"477a4977788c3eb13a8be61833ad9808e9055aa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:51:00.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6610c8e16a1954cdf685d57173704155b1c3e2bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 00:13:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cff10f1ceb34db7b58b5c0fdba4ab0a : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4534c001738d49c0b1de5792c079ceff : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0aad2b377414947a39ef064080c035d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d9a02fdcb64fa3bbfdbe7645d8a4c9 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/166dd81def6f40b99a1ef58a655020e8 : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/214cbf4dae7346dd891d72b6130b071a : SUCCESS in 1h 15m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c674d0596640a8bbe8ad5a3ef98a60 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/11e9361fb81f4411a38e9d67ef2b93ba : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":9},{"id":"33dbb726ad8ca7cd0c8119479338005b559eeccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 00:13:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"5c0368eefab6c28c5d905b146eae93af75876384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 00:14:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76181748222481cad08c1a971a7f981 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd8c7148cffb4ef6b3eed24915c4e3d0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"41e6b9bed8e04a21ad9f81057aefce79a70c0aa3","revisions":{"1c38ebbe8026201c667d4859fc6614a95272e076":{"kind":"REWORK","_number":1,"created":"2020-07-16 13:48:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/1"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-16 13:47:07.000000000","tz":120},"subject":"WIP: Add placement client methods for allocations","message":"WIP: Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c38ebbe8026201c667d4859fc6614a95272e076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1c38ebbe8026201c667d4859fc6614a95272e076"}]},"branch":"refs/heads/master"},"6b2186dd9429db26b21cf8297f8008f2fa09d26f":{"kind":"REWORK","_number":2,"created":"2020-07-17 08:49:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/2"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-17 08:49:26.000000000","tz":0},"subject":"WIP: Add placement client methods for allocations","message":"WIP: Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b2186dd9429db26b21cf8297f8008f2fa09d26f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6b2186dd9429db26b21cf8297f8008f2fa09d26f"}]},"branch":"refs/heads/master"},"7fe5c3de40f261c5d2bd74c5164885d9d5edcdf5":{"kind":"REWORK","_number":3,"created":"2020-08-03 15:12:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/3"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-03 15:11:34.000000000","tz":120},"subject":"WIP: Add placement client methods for allocations","message":"WIP: Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7fe5c3de40f261c5d2bd74c5164885d9d5edcdf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7fe5c3de40f261c5d2bd74c5164885d9d5edcdf5"}]},"branch":"refs/heads/master"},"f7f8c54d9878c994a2ccd43ed3b5f3832b5287f5":{"kind":"REWORK","_number":4,"created":"2020-08-04 10:39:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/4"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-04 10:39:25.000000000","tz":120},"subject":"WIP: Add placement client methods for allocations","message":"WIP: Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7f8c54d9878c994a2ccd43ed3b5f3832b5287f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7f8c54d9878c994a2ccd43ed3b5f3832b5287f5"}]},"branch":"refs/heads/master"},"01cea64276835e2454f87c93fdad67b37e7cff82":{"kind":"REWORK","_number":5,"created":"2020-08-07 09:47:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/5"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 09:47:08.000000000","tz":120},"subject":"Add placement client methods for allocations","message":"Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01cea64276835e2454f87c93fdad67b37e7cff82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01cea64276835e2454f87c93fdad67b37e7cff82"}]},"branch":"refs/heads/master"},"737b2c34acba480fb41e44c58e937bcf463f56a0":{"kind":"REWORK","_number":6,"created":"2020-09-03 13:48:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/6"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-09-03 13:48:54.000000000","tz":0},"subject":"Add placement client methods for allocations","message":"Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/737b2c34acba480fb41e44c58e937bcf463f56a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/737b2c34acba480fb41e44c58e937bcf463f56a0"}]},"branch":"refs/heads/master"},"45f166cb8cc757be855a64dacf115290e09cfc5d":{"kind":"REWORK","_number":7,"created":"2020-09-03 14:23:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/741452/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/7"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-09-03 14:23:56.000000000","tz":0},"subject":"Add placement client methods for allocations","message":"Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/45f166cb8cc757be855a64dacf115290e09cfc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/45f166cb8cc757be855a64dacf115290e09cfc5d"}]},"branch":"refs/heads/master"},"39790e102365e16a5dd8eee920231c9415771c0f":{"kind":"REWORK","_number":8,"created":"2020-09-03 20:09:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/741452/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/8"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-03 20:08:41.000000000","tz":120},"subject":"Add placement client methods for allocations","message":"Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations\nand for updating QoS minumum bandwidth allocation.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/39790e102365e16a5dd8eee920231c9415771c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/39790e102365e16a5dd8eee920231c9415771c0f"}]},"branch":"refs/heads/master"},"41e6b9bed8e04a21ad9f81057aefce79a70c0aa3":{"kind":"REWORK","_number":9,"created":"2020-09-03 20:13:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/741452/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/741452/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/741452/9"}}},"commit":{"parents":[{"commit":"3d6e16de32a41737e620802a888027ab2a949197","subject":"[FWaaS] Add note about deprecation of the neutron-fwaas project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3d6e16de32a41737e620802a888027ab2a949197"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-10 15:57:36.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-09-03 20:12:58.000000000","tz":0},"subject":"Add placement client methods for allocations","message":"Add placement client methods for allocations\n\nAdded placement client methods for listing and updating allocations\nand for updating QoS minumum bandwidth allocation.\n\nChange-Id: I9dab2e9e9728183ee209a038a0347eb7b4a83a1c\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/41e6b9bed8e04a21ad9f81057aefce79a70c0aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/41e6b9bed8e04a21ad9f81057aefce79a70c0aa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
