)]}'
{"id":"openstack%2Fneutron-lib~746552","triplet_id":"openstack%2Fneutron-lib~master~Ib83f30696269c968f41f9c1131596268a4412658","project":"openstack/neutron-lib","branch":"master","topic":"pylint","hashtags":[],"change_id":"Ib83f30696269c968f41f9c1131596268a4412658","subject":"Start running pylint for pep8 tests","status":"MERGED","created":"2020-08-17 15:57:03.000000000","updated":"2020-11-10 02:03:52.000000000","submitted":"2020-11-10 02:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746552-1604973739007-0535b842","meta_rev_id":"5418381a505f6ee8b4f4a0a78ffb306e46565770","_number":746552,"virtual_id_number":746552,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-10-28 17:56:58.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 02:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2020-11-10 00:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-09-18 07:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-18 07:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2020-11-10 00:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 07:51:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-18 07:57:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-11-10 00:27:32.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-11-10 02:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f277bdc4a99a9f34c5cd811f0506ff16db14dbb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-17 15:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0d39bcf117f36e646967f8d21e11e2560d878bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:21:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e081c4b959642d3a286bc5172ebf449 : SUCCESS in 7m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92be55a216ce4bd79f2adc0152b4cf68 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc172f9fc8f4b31b98fbab89c4a54ce : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69b3d0d8cf840d58fb1c42c2106cd82 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd6e1755aadf42c1a4de3f1456af1738 : FAILURE in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a696dd10844147bb91eecc80e815a0a4 : FAILURE in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be1d960d317b44e3881a84f3272cb9e9 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ce66071be841f794bb2253e180fae9 : SUCCESS in 5m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7dff9ede4a7410cb5987e89862b1ff0 : SUCCESS in 1h 17m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6391fcf90b480b9bb5dd63ef115e97 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"440d56df01ace6baa60bcb79c853cdd022a6f727","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-17 18:46:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"734767f573f5fd830954e22712b610c27538343e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 20:21:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/356badb19d8e4b70a386d3f3a765b8e9 : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0577b08c73a1453e837753422c2d5878 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81b50062f224208bb892190f1ec3b11 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad0a89ec83b24d168d419295f85fe77a : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/378dfc8a7da74a5fbd537ce66fc65ea5 : FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e80efd8324e4e1ab2c8a38bfd4c93d2 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61eb4fc6297b44029dfc12166d1641f0 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b6b2e4be94f48e5bb2204827090a44b : SUCCESS in 5m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6b2b5e4069a488fabc800046fcca97b : SUCCESS in 1h 28m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4e0186ba2af4dbab230587597a5587a : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7736361521d428bca610beb643764f13a9ac0cc7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-18 19:13:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e6a4b1fa602e549feabd0dc9eccc81b87738c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 19:48:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7e9fd02882a49ecb4c0158dcf152c4f : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ff84ef6bc974b0bb7f55bd5f7e3479f : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb8d8266b9c423abb45f54988c999ca : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b87ddad8a40492da39ffe9e7eaea113 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f97d3ce9a5814af6880a13bf6855d5ef : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6b339fa5a7436396b7e28d05501e23 : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/630d247bcbe14efa96d9f42fce420cf6 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2747945c89754796a906167c3ae36ef5 : SUCCESS in 4m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e46a0517494e470e97bcaa75dbb7abac : FAILURE in 28m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c748d7111e941edbdf9c8da4f74dbff : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c96cec3e3ab22c08f8fc6eebd492279d5c0348ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-18 19:48:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b87b8615febe4f2d760bdc55a4d3cdfc932d7ef2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-18 19:51:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2a13ee768615a8309211fc3500238d3be1f1c5db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-18 20:26:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"73c8f6794a745357990fc6f764d02707ece23129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 20:55:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67a1cdc22aa04defab9ea22f88022a43 : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f1a36fa2ca14bdf868d817a0d88cbd6 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1108be8a8d694a93b1d98bfa1b822d03 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d103d0e254c1421ea2b7bb42bf29effd : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2756536c28c410d84b7691d401cb006 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5303c8b36dd645759f2bdcdb038990a3 : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5261e2efb784468692ae3f64c8d29ca9 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae2816e1da174d8abcde322a2662e7f7 : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cddcf592125a48faa1708284dc90df7f : FAILURE in 28m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d3291e2cfa406eb81c7e42420f9859 : SUCCESS in 4m 15s\n\nWarning:\n  Comments left for invalid file neutron_lib/services/qos/base.py","accounts_in_message":[],"_revision_number":6},{"id":"e1070362ff31972dcef664a455f53e780d81f57f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-18 20:57:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"20e139e8ecc787bd3abd30d811967f08bee518db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:30:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49b07197070f4dc3849b121033f5b4eb : SUCCESS in 4m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8aca1e3e89d943d0a2151307ada9f417 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8340470bc2484bfe9e469e15262a34fe : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341a5783aca8431194978b71a56fd6e4 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/345d332b2923434d8ab299bc6e5ac095 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74b910f5e984688bd30d15048818a19 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e94326f6174c6dbbf57f68af72ec8d : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7510cff097754364b36dfe9ece07a40b : SUCCESS in 4m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c00b367ddba846aa93d6f752adb1468b : FAILURE in 31m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b131f2b796d407d89771788f775eab9 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":7},{"id":"14f2979e7bd28611664f566282ac1cec2bc763b6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-19 11:15:45.000000000","message":"Patch Set 7: Review-Priority-1\n\nI\u0027m not exactly sure why it\u0027s failing but it seems for me that tempest-full-py3 job\u0027s failure is related to this patch. See neutron server logs from that job: https://5a56c27bc6d0e55a7596-de61bf74e5db0c358b811211095910b1.ssl.cf5.rackcdn.com/746552/7/check/tempest-full-py3/c00b367/controller/logs/screen-q-svc.txt","accounts_in_message":[],"_revision_number":7},{"id":"64d792e5872c2fe3284ce15003978a88cfa2c34f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 12:12:42.000000000","message":"Patch Set 7:\n\nYeah, I probably broke something.  I found this when cleaning-out repos so not a priority.","accounts_in_message":[],"_revision_number":7},{"id":"27d68d482ad7b4749ab13fba720ff8213352d4fd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-20 21:07:28.000000000","message":"Patch Set 7: Code-Review-1 -Review-Priority","accounts_in_message":[],"_revision_number":7},{"id":"547040f1f75ab16a8dc6bd6b4e80fcaf788c6bb1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-28 17:16:30.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"88a8198c0f9e5e19b76b1a63e6557fe19da03446","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 14:24:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7bd682202a10c104eb8ad143e2a15df81f6cb40b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 14:24:46.000000000","message":"Patch Set 8: Workflow-1\n\nTrying to determine which change is breaking things","accounts_in_message":[],"_revision_number":8},{"id":"3b70e9588f2e4adc0a1402f5056ea9ea198b8f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:20:36.000000000","message":"Patch Set 8: Verified-1\n\n(21 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30cd9694f1cf40e8878eb9704228a796 : SUCCESS in 4m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa57b078d12747d6865fff02b2ede17e : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb98e2094e744c7f9227f26ef482eba8 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06ff1dc6e7c94236800392b8abd373ea : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ca560817a3d41cf9a152f050671c871 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905a4f0f8d154e1496263c287d3e9608 : FAILURE in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3125c0f5bfa14da495525fab86bdab93 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a716c8ed7c7c46dc825356148f9c3050 : SUCCESS in 4m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae2f50f1d5034b849600780d872476a8 : FAILURE in 27m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ad6a10c53f4e8c91705bb3b9f0829a : SUCCESS in 5m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/91f9c044f4a7491ea4b472d959dae524 : FAILURE in 51m 05s\n\nWarning:\n  Comments left for invalid file neutron_lib/rpc.py\n  Comments left for invalid file neutron_lib/fixture.py\n  Comments left for invalid file neutron_lib/context.py\n  Comments left for invalid file neutron_lib/exceptions/__init__.py\n  Comments left for invalid file neutron_lib/exceptions/l3.py\n  Comments left for invalid file neutron_lib/db/utils.py\n  Comments left for invalid file neutron_lib/db/standard_attr.py\n  Comments left for invalid file neutron_lib/api/converters.py\n  Comments left for invalid file neutron_lib/callbacks/events.py\n  Comments left for invalid file neutron_lib/plugins/utils.py\n  Comments left for invalid file neutron_lib/placement/utils.py\n  Comments left for invalid file neutron_lib/objects/common_types.py\n  Comments left for invalid file neutron_lib/objects/utils.py\n  Comments left for invalid file neutron_lib/objects/exceptions.py\n  Comments left for invalid file neutron_lib/objects/logapi/event_types.py","accounts_in_message":[],"_revision_number":8},{"id":"12f2f950e8f63d26d5421c7b267612a037fd61db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:18:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a369fb8158a51fb6bc2a9c726445ff275ac00c13","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:18:22.000000000","message":"Patch Set 9: Workflow-1\n\nStill debugging","accounts_in_message":[],"_revision_number":9},{"id":"60fc0c8baa5a5d7b2e9189b1ce6a49517330b8f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:52:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bc37809ea476b2ecaea1f0a301245b61b836a634","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 16:54:25.000000000","message":"Patch Set 10: Workflow-1\n\nstill debugging, but fixed new 2.6.0 errors","accounts_in_message":[],"_revision_number":10},{"id":"55cf921187acdbeb1e8197ef6b3853205351baad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 17:09:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"757b51f10eafab1c120894dbd11f4d8c15dcbb1d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 17:09:22.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"d12b8134e0e0246fde4dddec081abefa778b66ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:41:41.000000000","message":"Patch Set 11: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/decb927482aa437095e1ee29834dbe1f : SUCCESS in 3m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6e13fe94f054fd185f34991cd492d36 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fad1b303c44d03879ba6678e640b3b : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2706299a184842fba8d46cd834dd7289 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a40c0a664ff4b5ba62f6244ab7116cc : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab2c07efdf5474ea100d66367bf226e : FAILURE in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4977211eecc145a5bf1096f9d0fab146 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5da93e7777f465d94c8329f35448584 : SUCCESS in 5m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ce79549a68940b08a2ff04d2522839e : SUCCESS in 1h 27m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800fa9e1759c4db395b684a0b0f13fe2 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05f58685b014482c81c350c8a9e71de2 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":11},{"id":"9a342cdab8a1936dc1888f534cecf38ed0b2e2b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 21:17:40.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"097c2e06ead47a7c6261064e4a39f305da05f021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 22:57:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cea36324e708493ba005df0dc80f7115 : SUCCESS in 4m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8a06df4b9cd48a08da6a15290a12f86 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929f9a04b54441a98b150281637fb2b2 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c467d1e088a94cdf91757a8248701e2e : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0283860524d3421fa5f3508273b64caa : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea3ebcc5ea340d58b82ac28f3832c49 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9c93517ae0748b1a874b84829671fb9 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04c12e411a214368b28c4aacbe2ab10f : SUCCESS in 5m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de68cb78bee0464997c9b4b109aab224 : SUCCESS in 1h 18m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ef469c248f04208b829ab9011c3f298 : SUCCESS in 5m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2067832dbab0462fac3003d624ec2396 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":12},{"id":"10af480fb106d0d08e1c5084ca9b04fa3f0fb9b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 08:20:27.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"b456667ae4b879d5d3411da316a63df07a8fe4ac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-07 10:09:36.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7105be8e5153c1f91970b610bd2a5c79a4f741ea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-08 18:13:00.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7b64151936a789a35d5a998f78045299d1532daf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 08:08:52.000000000","message":"Patch Set 12:\n\n\u003e (1 comment)\n\nThanks","accounts_in_message":[],"_revision_number":12},{"id":"ac117b5d7592a44d4d7364ac8ea838b271b4d353","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-16 20:31:41.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"d7b048d5ac4f9adf7c1d5a3ceae34ee3fafea657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:31:53.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"d03bfdb9339e299e163181ec61bbd662b289569a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:11:04.000000000","message":"Patch Set 12: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a364f91e24d4f5f95a6a1fca5805f6a : SUCCESS in 4m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c2fb60f55e848419fbdaff34100c4c1 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda5f0aabaad49ec97e2335a69c318ec : SUCCESS in 17m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65879cd84b904efe974dae792ce14071 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4190869477984015b7d4ba38e96f1018 : FAILURE in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a886698f7c14acd8f2dcd7c86562b55 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7e73e39c3204ee7a7db5f05af2bcd27 : SUCCESS in 4m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4ea01991c9a4cbeb3fd00e8695747f7 : SUCCESS in 1h 36m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29b8ad79e5f147319a7d48aa54cfdffc : SUCCESS in 4m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/266b430ec193437c8ee0bbcd01f8742a : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":12},{"id":"3203982ea3b4a701da14c5893b30648d38be27b8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-17 13:51:57.000000000","message":"Patch Set 12:\n\nLooks like something snuck in recently, will rebase and fix it:\n\nneutron_lib/placement/client.py:783:16: R1724: \u001b[1;3;35mUnnecessary \"else\" after \"continue\"\u001b[0m (\u001b[1;3;35mno-else-continue\u001b[0m)","accounts_in_message":[],"_revision_number":12},{"id":"42bd9229c98d51c627494caf4a6884da19fa353f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-17 13:53:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1922e7c3d6c21b19306a6e3379ef0a80491b0b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 15:38:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aae6295171f84606b015047a3b674664 : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/961e49e5e3a347ec86cc75b88fd0d20a : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276bd4971fda472ca6a22e6fdfe003b4 : SUCCESS in 13m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1d5e5e1894246cd993b909e0ecb97a8 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c84a01fae1044e149a82bdc5cbe54714 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696b326e7784484ba41a2d64af566943 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1935c3cc0148db914910f2b174309c : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b462fd409a0a42cbbb45195e4c8a5dc3 : SUCCESS in 4m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d042dd59a514efa96e5f9c7d409a3a1 : SUCCESS in 1h 30m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97882671e4b49ea9eb7703906567977 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7647bc685113460f8ac5086dea284aa0 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":13},{"id":"8d5f4a83f9d1833aae7fc3ef1811e693b73b3e0e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-18 07:51:10.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"5bea0593656eff10c3bed7509f55d34378145873","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-18 07:57:14.000000000","message":"Patch Set 13: Code-Review+2\n\n\u003e Looks like something snuck in recently, will rebase and fix it:\n \u003e \n \u003e neutron_lib/placement/client.py:783:16: R1724: \u001b[1;3;35mUnnecessary\n \u003e \"else\" after \"continue\"\u001b[0m (\u001b[1;3;35mno-else-continue\u001b[0m)\n\nhmmm, that was me, sorry for that.","accounts_in_message":[],"_revision_number":13},{"id":"2ddd63f0141ee93608224ea021b445d7f4f10515","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-27 21:09:31.000000000","message":"Patch Set 13:\n\nrecheck\n\nthis can probably merge now","accounts_in_message":[],"_revision_number":13},{"id":"8ff1f4ccd694414c3fab1d0a7189dee54ac01937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 23:28:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4a88e8220323411fb4b45c3f73c411b0 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28d957a236c24861a3cb156f0d52f92a : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908286cb5b7d497d9e459b28ea743d97 : TIMED_OUT in 31m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/320812e002544c23ab77393686f1f6b6 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639eb3a93b0842d5bcb900efeab826b3 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ab4edabdae43048289af56bcfbc0aa : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c327c6dba60b4c298a8743bb0fb8638a : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6effca8c60fd46a89a9a6c3b8f5d102c : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e662a4e6a0f40ccb72f98709e3ef9d9 : SUCCESS in 2h 04m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/208e60fd55ce41cdb274d823cdccef65 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/792dfdd057004477ba30e42eb3f200e4 : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":13},{"id":"4ee0c935cacbd6773b20ac120a1de254459d360e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-28 16:48:32.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"2d7bff812d6f2f2577c78fbb02379afdfc641988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 17:15:37.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/878685c9baf04eb28501d0ecde2b02da : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/927fe7204abc4dd99196bf9b6b3020d8 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db61227a6e2c4f29b309c57df0837e0b : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6569f12d691649a9b11902c61422cab8 : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89d68fc7edb14ac4add7cb856b9cbfb4 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25937e3ae99466aa2b51a2e2fcebfd0 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff4c3cec750948a0bf0119f941c49fe3 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b62b805a8a0443ab574fff165691adb : SUCCESS in 4m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c743fbe0365649a2af2688e4035c841f : FAILURE in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca840e230d7a49fe8814de21a4da5150 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bafb21e6ab784e968cfecbec95592615 : FAILURE in 12m 21s","accounts_in_message":[],"_revision_number":13},{"id":"f2d87fe9005d859298dc27e97b9bc5592b535a60","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-28 17:56:58.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"087f1b3335785213ee76cb48bc51f10f48de9e84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 19:21:30.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b6fcb16ceee49dfa0403b447f7fe9ad : SUCCESS in 4m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d1445ea95fbc482cb99207a6dfbc8c05 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b08655d9f374d31a988f6e98762b6cb : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc304c2d8114642b41fd6fab84ed1e2 : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/466b7d6278b347129bca9fd7b0fe61be : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba52b77c1268422bbc98a3d7924d3e61 : RETRY_LIMIT in 1m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d181f892d844dfbd54a07b08679263 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9895b31fcabe4490afb55dc7e87ccf80 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78e299673e0a4c56a16cc17e834feb9d : SUCCESS in 1h 15m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/592de6b06c384e3fa1f1ba21289eeaef : SUCCESS in 3m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fd035144aa3943a1ba8e29669f979530 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":13},{"id":"2f9b8d29401101178c21e379c13c631cbfea2266","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-02 09:54:57.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"63b1a9115159d41a5828cda982b7ed3063fc0013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 11:16:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93b1dc0b53a3406caf1be67c72160ef6 : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6454f5b17f1400aa00d6f1c2a2ee7cd : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778eb7cb4e9249879b17e74e7085e5f4 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3834cc1245014be6bbae7503c9e079ec : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301261e022f448ddaae95b7edcd89377 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07db0b3097e24eca91e9e3210dffc545 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32c94972cc8447c89e41e6d73f73d88b : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ac6300a3766484f8fbab9a6db262cc1 : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5514e2b8f216460683c14fd615e827e6 : SUCCESS in 1h 09m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53eb195eef6641829ad6dcb175cc5cf7 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1e212576b99a4dd1801e1682238ebe2d : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":13},{"id":"81b203154d74fec4e0d56935048d96fb38d2d933","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-11-10 00:27:32.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"62fb70a9b45e7f024719e2bdd2dca6d229a6aa5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 00:27:48.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"a301d03e7846547d1dfe0d0fbcf6af3a92876550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:02:18.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e8eb72e475cd412c87375f46b8451b90 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/27471cbeeec44112a7953a7eed878a8e : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7df5cbb5b7d435d9de969d24d9f9281 : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb79e50be36f44ad97886188c31605d3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6bb51ff4a5493cb09711472ad49a61 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daac0ef7e01c45f7ab4e72f33a1f1ad9 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/939f4b0fcbca4fa083ddc619852047e9 : SUCCESS in 4m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/217a0cd1bf2740a79218b450a4b39eff : SUCCESS in 1h 30m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3f40738b2c346de92a324b3dd220353 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f97a8f1599da4ce6bd0f8230274bdc22 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":13},{"id":"43f5f99a24a5928934ed04c72cac3068c1332530","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:02:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"51063283461b92dd019d5d79eca0ee66e5b2acc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:03:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2afbe3d78c1545989e5a5c17b5bd8322 : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7d82b47ebb479caeaaad325354c621 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6ab98a58b414f9183eab200b3f704ff : SUCCESS in 51s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"e92e98f8c63302d49adc90be92a0cd8a9ceaa4b1","revisions":{"e7e68cb52220420f8f2c2357b540a3cdaaf778c4":{"kind":"REWORK","_number":1,"created":"2020-08-17 15:57:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/1"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-17 15:54:40.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7e68cb52220420f8f2c2357b540a3cdaaf778c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7e68cb52220420f8f2c2357b540a3cdaaf778c4"}]},"branch":"refs/heads/master"},"c037b68078945ae19b5501da1f035184d6eb9d6f":{"kind":"REWORK","_number":2,"created":"2020-08-17 18:46:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/2"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-17 18:46:06.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c037b68078945ae19b5501da1f035184d6eb9d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c037b68078945ae19b5501da1f035184d6eb9d6f"}]},"branch":"refs/heads/master"},"11829e3f23a56b5236aa5c8221d78bc947819dd1":{"kind":"REWORK","_number":3,"created":"2020-08-18 19:13:09.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/3"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-18 19:12:13.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/11829e3f23a56b5236aa5c8221d78bc947819dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/11829e3f23a56b5236aa5c8221d78bc947819dd1"}]},"branch":"refs/heads/master"},"172274285982af62eb7d823c86750d4ff88032d7":{"kind":"REWORK","_number":4,"created":"2020-08-18 19:48:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/4"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-18 19:48:14.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/172274285982af62eb7d823c86750d4ff88032d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/172274285982af62eb7d823c86750d4ff88032d7"}]},"branch":"refs/heads/master"},"e7731bde1e1aab0bdac276a42e4b21f2f095c297":{"kind":"REWORK","_number":5,"created":"2020-08-18 19:51:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/5"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-18 19:51:41.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7731bde1e1aab0bdac276a42e4b21f2f095c297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e7731bde1e1aab0bdac276a42e4b21f2f095c297"}]},"branch":"refs/heads/master"},"7bdaea3105d90e725be39a9af52f71c16e6e881a":{"kind":"REWORK","_number":6,"created":"2020-08-18 20:26:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/6"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-18 20:26:05.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7bdaea3105d90e725be39a9af52f71c16e6e881a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7bdaea3105d90e725be39a9af52f71c16e6e881a"}]},"branch":"refs/heads/master"},"856962805f5d5914f13da5e221ebbadfa031d083":{"kind":"REWORK","_number":7,"created":"2020-08-18 20:57:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/7"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-08-18 20:57:14.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/856962805f5d5914f13da5e221ebbadfa031d083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/856962805f5d5914f13da5e221ebbadfa031d083"}]},"branch":"refs/heads/master"},"6927f813b507a15735789a8ce5e65d431ea3201c":{"kind":"REWORK","_number":8,"created":"2020-09-03 14:24:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/8"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 14:24:15.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6927f813b507a15735789a8ce5e65d431ea3201c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6927f813b507a15735789a8ce5e65d431ea3201c"}]},"branch":"refs/heads/master"},"36abd17ed2b69ed5e1617147b418f638319f63cb":{"kind":"REWORK","_number":9,"created":"2020-09-03 16:18:09.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/9"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 16:17:55.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found.\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/36abd17ed2b69ed5e1617147b418f638319f63cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/36abd17ed2b69ed5e1617147b418f638319f63cb"}]},"branch":"refs/heads/master"},"06ae469b261ba4a3937cd80578fd8fd7f25da045":{"kind":"REWORK","_number":10,"created":"2020-09-03 16:52:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/10"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 16:50:14.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found, 2.6.0 added two new checkers\n[1] that make pep8 validation fail:\n\n- raise-missing-from (W0707)\n- super-with-arguments (R1725)\n\n[1] http://pylint.pycqa.org/en/latest/technical_reference/features.html\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06ae469b261ba4a3937cd80578fd8fd7f25da045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06ae469b261ba4a3937cd80578fd8fd7f25da045"}]},"branch":"refs/heads/master"},"080d7aaee5c5180767368b75b7255d064c32ef53":{"kind":"REWORK","_number":11,"created":"2020-09-03 17:09:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/11"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 17:09:06.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found, 2.6.0 added two new checkers\n[1] that make pep8 validation fail:\n\n- raise-missing-from (W0707)\n- super-with-arguments (R1725)\n\n[1] http://pylint.pycqa.org/en/latest/technical_reference/features.html\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/080d7aaee5c5180767368b75b7255d064c32ef53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/080d7aaee5c5180767368b75b7255d064c32ef53"}]},"branch":"refs/heads/master"},"718323c295cff8040b7a4965e29fc48b46c5e39c":{"kind":"REWORK","_number":12,"created":"2020-09-03 21:17:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/12"}}},"commit":{"parents":[{"commit":"96e1d028b84419d187f085b587e672447df00ae3","subject":"Merge \"[api-ref] Added warning about overwriting router\u0027s default routes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96e1d028b84419d187f085b587e672447df00ae3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-03 21:17:03.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found, 2.6.0 added two new checkers\n[1] that make pep8 validation fail:\n\n- raise-missing-from (W0707)\n- super-with-arguments (R1725)\n\n[1] http://pylint.pycqa.org/en/latest/technical_reference/features.html\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718323c295cff8040b7a4965e29fc48b46c5e39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/718323c295cff8040b7a4965e29fc48b46c5e39c"}]},"branch":"refs/heads/master"},"e92e98f8c63302d49adc90be92a0cd8a9ceaa4b1":{"kind":"REWORK","_number":13,"created":"2020-09-17 13:53:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/746552/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/52/746552/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/52/746552/13"}}},"commit":{"parents":[{"commit":"18129456d8b6092ffa96de0f9a12c6156d1cf5d8","subject":"Fix lower-constraints job for focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/18129456d8b6092ffa96de0f9a12c6156d1cf5d8"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 22:41:21.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-09-17 13:53:01.000000000","tz":-240},"subject":"Start running pylint for pep8 tests","message":"Start running pylint for pep8 tests\n\nWhen \u0027tox -e pep8\u0027 was run it didn\u0027t run any pylint\nchecks, let\u0027s start doing it.\n\nFixed issues pylint found, 2.6.0 added two new checkers\n[1] that make pep8 validation fail:\n\n- raise-missing-from (W0707)\n- super-with-arguments (R1725)\n\n[1] http://pylint.pycqa.org/en/latest/technical_reference/features.html\n\nChange-Id: Ib83f30696269c968f41f9c1131596268a4412658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e92e98f8c63302d49adc90be92a0cd8a9ceaa4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e92e98f8c63302d49adc90be92a0cd8a9ceaa4b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
