)]}'
{"id":"openstack%2Fneutron-lib~747523","triplet_id":"openstack%2Fneutron-lib~master~I26f7e642214bd09a6d4608c9cfdf8c5ef2537489","project":"openstack/neutron-lib","branch":"master","topic":"bug/1877301","hashtags":[],"change_id":"I26f7e642214bd09a6d4608c9cfdf8c5ef2537489","subject":"Adds l3-ndp-proxy extension api definition","status":"MERGED","created":"2020-08-22 09:28:05.000000000","updated":"2021-07-29 09:33:13.000000000","submitted":"2021-07-29 09:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"747523-bug/1877301","meta_rev_id":"be2b9cd0a68949edb0f812b0c92d78ea415a9f78","_number":747523,"virtual_id_number":747523,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 09:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-29 08:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":2,"date":"2021-07-28 04:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-07-29 08:04:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 07:05:40.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-08-24 10:46:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-25 13:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 15:01:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-22 19:35:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-26 00:59:12.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2021-07-26 01:06:05.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"f1d7277dd58ccf4450dc2b3d1b88d971d33838c1","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-08-22 09:28:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2f73c5d306532faf3b8d7c584ee3a687551f131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 10:55:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6a41d930a04c3a99546e8e07770aa0 : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2fbe9ccea88417899d306e1342d1a08 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8703b6350cc84963ac5833794e1d5bdb : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2562611b9bb4ec39b0257c8c3b9c75f : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c12dd5eaf964e3fab1a7d79cebd1202 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/429a9e4aea7944dba96c9c5866b82cd8 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e253569776f414da216e39dbe2b63e8 : SUCCESS in 1h 22m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78ac296c0dc74109845036b501e552cd : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0820fe69d258d344d5139f777ae4417127689ad2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-24 07:05:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f8b20ee3c9950c4b1f8b5d23b738dfa93e46575","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-24 08:33:56.000000000","message":"Patch Set 1:\n\nMarking it as WIP just to make sure we will not merge it before spec https://review.opendev.org/#/c/728628/17/specs/victoria/l3-router-support-ndp-proxy.rst will be approved","accounts_in_message":[],"_revision_number":1},{"id":"e8eb267738933804f9336c5153a284f8166c4ef6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-24 10:46:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fa08d01901d5841d7a5436022b3d087b8f69084d","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-08-25 12:05:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"945e544bff418e0ebbfcef1f907bab0d5ac09c2c","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-08-25 12:05:55.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2a51be37d6677d764f50aa752700ae05fd660312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:24:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5fa5e02094435aad36b9c60a1b35b3 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42827591943b4100940f6b561ed6b482 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be7487030fcc4d579640418794f98eff : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc600feb9154b958005f71b00d8fca1 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c7ee9132dd4c85a745f12ad1684250 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f56658f2d034412b9ec937cb4bd6e1c3 : SUCCESS in 5m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffd2abc8b4e943b782ec9a8e8682af55 : SUCCESS in 1h 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d94527a2f124453b006ba5f6180e665 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e2d3f104fe8657b7ebbedf94a5b2841b22959041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-06-29 09:26:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"658ad7978da187a6df2efe1cb1e0b55379d8b423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 10:54:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891da25e62644e1a9ef528717d86bc45 : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d23740d2f73343768615b4edac960e8d : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14021fbe70e245e2bd89daffeb09eb6a : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cf7f2f080245a7930c5a44c6fd1493 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b7d0afa5b044787b10b7c2b3230e75e : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f9c16317044510b6d5fa4edeb65499 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbbae548a29d4fc1866bb050f8007aef : SUCCESS in 8m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6fb35c4bff345dda4ce971a78acb0cf : SUCCESS in 1h 26m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf52246ad87148018c0ba187135840bb : SUCCESS in 6m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/31bdd196b4c441b4a4cf29f795d06051 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2fe1257e8ddd3c46e685511fef04ff572076eed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-17 08:10:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6fa5e7c25da301b49f17c7412befbbd58d5a8e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-17 08:14:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"215e3514d419d995e40c12249eb3746538e4d46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 09:26:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78766ac9c8464b2d9e929a024f783d48 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1898fb5ef3b46c8af9cd786475347e2 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9101b4c698e44451b5b0b75edeba664d : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd37799f1844b83b3cd52af41c28b4e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b943d1da9f9b4a5b837bbb1a121def72 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d2bda2cb0b42b08d949cfd1837f155 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/946a1f4ebf4644a8aefcb887e0c1b48d : SUCCESS in 4m 34s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f3148eb43294fea8ea8e2cbb2022a34 : SUCCESS in 1h 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3f2b7d1e3014ab8a8ba2394bacf6f9f : SUCCESS in 3m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e97b967467a4c4c842503f458d5b53d : SUCCESS in 41m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c01924a975c34bbc80b6e8de175d1447 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":5},{"id":"3b29da235c0858307cf83c47407ab132b52a14af","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 13:12:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"19ccf32af61ef07e0d3098a32a4dc29dc228232f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-21 10:43:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"75df18d3f20155e3deb4265b6477c78f5dff8755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 11:26:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c15ac780924b4eb64be08f52d61a71 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8edb82ce0564929ab14c9a62ac75b01 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9a3928790744dd68c400cf5a0a442bc : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849deded1c65402c970c7dcf4ae2df14 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1089418904f34f13976b7835f5fb7431 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e60424fd887d40658a3fb211f2e25c1d : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d88da1e42ca246b2a6555d6aefd55142 : SUCCESS in 5m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/461b11c661dc41c1b85634af82d15e5c : FAILURE in 20m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcb99b94537542aa8c06e4b07850c1b5 : SUCCESS in 4m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9af40693419542e09390794c41944a4f : SUCCESS in 32m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81354c8e66fe43519b24287aca88af02 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":6},{"id":"1910bca53b973fc4efc16ac43149c11d82b351ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-21 19:52:49.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"95d29273ea7337d949739d91552bbf0edc54a922","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-22 14:53:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"790872ec8b3d4a9a6e3ba0e9e2d8c74cde0b856d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 16:16:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74649b0c675b4de293c43de7e5f2212a : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/591ccb5c75be4b52a3dd59cd122db9e8 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a8dc68a78644e2f9c73f38771f2b852 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d32ee3d4abc439e831c046a2dddd89f : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab39ff5ae074226b1deacd485e45785 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b9ac20e12514f7ea7500b0acf4cf52b : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c427d3b43cb44ec7a71e0382eada1248 : SUCCESS in 5m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b2d1b44d68643edbe8ac1a5deefb807 : SUCCESS in 1h 21m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01fab3c5bedf4fdc9894cd3a85277e1a : SUCCESS in 4m 46s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ad8ec4d0fa54ae98609a1a007604e68 : SUCCESS in 38m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b10c09187e59411aa0ec235da143f50e : FAILURE in 46m 37s","accounts_in_message":[],"_revision_number":6},{"id":"f9f01a763a16d0352affe739a1aba5736b3454c8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-22 19:35:29.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)\n\nProbably not the best one to review since I haven\u0027t been involved, but I think I found some issues.","accounts_in_message":[],"_revision_number":6},{"id":"afd5901175f206fc201203c9a4a47655b7e4e676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-23 11:28:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6f8a666f51bc0d763692517c12b239c4164c39f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-23 11:33:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f6f3bf124c1e2b9f5766732b43568846fd22d322","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-23 11:35:09.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"65cf56fc0fab5f3045ac516c3a37a1c6af217066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 12:43:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc2bfd594384de59844139a346cbae7 : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bef4eb47ad404a8d86c8871dad4a5d84 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a1e2434db174798aed3a2ef4b02cb9c : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cfb09e7f20479fbf5dfff3420b6c9e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3e47eb226747a0903dc7b24c03fed2 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220d781765c8455d9de9a5e884988c31 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b54329ee0e334963a6326299ad510ca5 : SUCCESS in 4m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6db521ea39b42d79cb7a5ec8cce9a63 : SUCCESS in 58m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee5136a6e3864252809fe785eac1a7dc : SUCCESS in 3m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc8313e94a974b748feaa15cc3423fe5 : SUCCESS in 35m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62b42bbdaae24dff9d6728e0766a5a7d : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":8},{"id":"0e1509222a3002cb5eea8433eb9411cd3e189113","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-23 13:26:21.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c8e7a4ddf72c39e9bef16578cfda4960c09baae4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-23 13:32:15.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a9173c1818ec1a9cb38ab854b7468be39d89d51","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-26 00:59:12.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e0b6ef82e4de94b6fdef7a35b780ce3fb5cf410b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-26 01:06:05.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"94cfd8db2489d1d17447705719c16e9cf97f1e51","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-26 10:57:01.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ae4f060fc5c037ae3fa70c233ae65f3815909666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-26 11:01:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"79d8315399d5cecad2df1806fd37ae196ade1017","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-26 11:06:44.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"79a1f3c6ac4be033bcedbdd5e50ab64d3fd8cf5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 12:18:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c743c493fe954c979286c533267e1712 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac30a6288e0543bd9c08196c896dfccb : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61e1247734fc469b9148b69a1cdada15 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6e0bd2f15a4c38a261623c60a24075 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f67430747814d60bde36396c3f91f4c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71d821f1bfb340e5b064f287ce4464a3 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6d6cedcef1d43dbb2e85f98aed9e7c9 : SUCCESS in 4m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8c24af0640c4053996460e67839b687 : SUCCESS in 1h 15m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2991bf6aec854ac7a01bed9b07d0c5ad : SUCCESS in 4m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ac25a1cc7a32461db4281d908ed4ba93 : SUCCESS in 39m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/452c2a5051da435080570f3032a65a70 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":9},{"id":"fc8ace595d44092a3ef71bbeebb07b10f5aee98b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-26 13:05:02.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2622e7aa948ebaacd9f42bfbbcc497bdb4960a57","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-26 13:22:07.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b04a643bfef84c6b652fdaab8fc26874c9d066a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 10:23:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2c3d2c3df9294d0271bd38bf74c68a5829c6ecfa","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 10:24:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2ec852b9c1c5f0aa29ac47af84e1d73e0a313508","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-27 10:50:00.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5308c87e10835fa4296aa057b995eb0204d9c8dc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-27 11:06:12.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e67fa663aa3e1d4509bed0ee69586340b3adf802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:52:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51230fa703b54fa3ad12c4642d574f3c : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b177cac0476844fd92f1e463549768e4 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf2e2b6a2ffd4f1bbf513dde4e8c50ba : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf46e7f53132441ea6b0bc5af991b315 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f168924f9af64df382e63c0ee261bb1f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12728b1de0442ceb6dd8694d2f3aad6 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4bc6da5f9134600976799b673cd692a : SUCCESS in 4m 57s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8edd4ed888548d08e2ff9583652b5e0 : SUCCESS in 1h 22m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/757c8e0a35cb4382ae82ad342ed20619 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c63fad0b31884724b333f6a4869cb2d1 : SUCCESS in 43m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/39f9987a5c1e429a87eec0b01899708d : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":10},{"id":"38ebd0cbc742ded5a2ead3102d95ae6d57ae3f3a","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 12:58:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9102d64b62ec322278709fe080e53a012a1f241d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 13:03:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"cd414faae7bf15d4f445895e350d73491b252a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 13:06:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c850c340a0ec12c6cd9929913edb79923942bb45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 13:11:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b099d9a28d8e8c02a82826cbbedee4391a409a79","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-07-27 13:23:50.000000000","message":"Patch Set 13:\n\nSorry for so many nits in my codes. A long day\u0027s work has dulled my brain.  I went over it carefully,  hope there are no issues left out.","accounts_in_message":[],"_revision_number":13},{"id":"bf5953905d68212eafc3c602a6705f11927dfadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 14:33:27.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18083ac3f90849a6aec859574af9b787 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0ae056322b2430a8bbb3af5e5284716 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30e7c54335f944c8a98c31d7ec3a58ef : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4c0f33277a4bc9980fd9dca346f2d5 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9783971ad740cfb2e4c85c9b2aefb5 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ce04849a089431a92e9fa267ee94430 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99c50156149143348ab9c62ba1add5c3 : SUCCESS in 5m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53ca53f764514de9b8ab7ff6db261205 : SUCCESS in 1h 19m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aec14945e0145988cc9caa99bf0ee0d : SUCCESS in 4m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/72e9e9c19aa249049ed4ebf2e49aae2e : SUCCESS in 40m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db4ee23950fa41f2bd73e905b2f6fbbe : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":13},{"id":"82a8661259fc7c1c2ad96b20c0647e9c350c46a1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-28 04:25:31.000000000","message":"Patch Set 13: Code-Review+2\n\nI don\u0027t think two separate API extensions are required but it is not a blocking issue. Looks good.","accounts_in_message":[],"_revision_number":13},{"id":"feaa84c57379849ed094f755ee38ba5377c8fb98","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-29 08:04:58.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"62e1100441d558f043eaee6ad55b641733bf3cbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:05:27.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"476007727949617874fb953b73f5b42c26602302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:31:43.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ce0f5d11364080bfcf4f7efbe18869 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9bc461310054fbfb8a394b62fd3e1d1 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e108df6df0e44ec28b72a4b5978de9da : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/466c45ab911849cfa1aed1e7c5c022e6 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9374947bcbf44d8a8810a77ce8484bfa : SUCCESS in 4m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59cf314c49384e05b70a2c7d06bc9175 : SUCCESS in 1h 20m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/134064cb935248a186b07f10cfc8fe5b : SUCCESS in 4m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3e9428d5306847c8b7e948394413802e : SUCCESS in 44m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a87b1fdc4b604890854875717ba2b4de : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":13},{"id":"0263ab91e39f75d371f90711affe1c06ece11f18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:31:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"be2b9cd0a68949edb0f812b0c92d78ea415a9f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:33:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd81e11d9c5a421fb048148038d05715 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ca25cfaec1542acbbad040542cebe62 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"2b46c1f74cbf076fca1a100a978a3a8a23d0338c","revisions":{"a722e4fb0aca3025e6dd94888149bf946881d9e6":{"kind":"REWORK","_number":1,"created":"2020-08-22 09:28:05.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/1"}}},"commit":{"parents":[{"commit":"40efa204df9feaf0d207b0e89eaa7a7776e02248","subject":"Fix: port attribute name propagate_uplink_status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/40efa204df9feaf0d207b0e89eaa7a7776e02248"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:27:47.000000000","tz":0},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a722e4fb0aca3025e6dd94888149bf946881d9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a722e4fb0aca3025e6dd94888149bf946881d9e6"}]},"branch":"refs/heads/master"},"e73c5ad3a4c7c62f28f439b2600431476e92a707":{"kind":"REWORK","_number":2,"created":"2020-08-25 12:05:38.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/2"}}},"commit":{"parents":[{"commit":"40efa204df9feaf0d207b0e89eaa7a7776e02248","subject":"Fix: port attribute name propagate_uplink_status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/40efa204df9feaf0d207b0e89eaa7a7776e02248"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-25 12:04:44.000000000","tz":0},"subject":"[WIP] Adds l3-ndp-proxy extension api definition","message":"[WIP] Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e73c5ad3a4c7c62f28f439b2600431476e92a707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e73c5ad3a4c7c62f28f439b2600431476e92a707"}]},"branch":"refs/heads/master"},"7d6f7176ac5341eec88b2e832c10b47061f5599d":{"kind":"REWORK","_number":3,"created":"2021-06-29 09:26:06.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/3"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-06-29 09:24:41.000000000","tz":-240},"subject":"[WIP] Adds l3-ndp-proxy extension api definition","message":"[WIP] Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d6f7176ac5341eec88b2e832c10b47061f5599d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d6f7176ac5341eec88b2e832c10b47061f5599d"}]},"branch":"refs/heads/master"},"cf6c504ae6a001eff8f0b89ee07439ba220e95fa":{"kind":"REWORK","_number":4,"created":"2021-07-17 08:10:23.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/4"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-07-17 07:48:06.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cf6c504ae6a001eff8f0b89ee07439ba220e95fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cf6c504ae6a001eff8f0b89ee07439ba220e95fa"}]},"branch":"refs/heads/master"},"e40368158bb4aee11d5bcc4f6227b5aedaaf7701":{"kind":"REWORK","_number":5,"created":"2021-07-17 08:14:37.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/5"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-17 08:14:24.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e40368158bb4aee11d5bcc4f6227b5aedaaf7701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e40368158bb4aee11d5bcc4f6227b5aedaaf7701"}]},"branch":"refs/heads/master"},"173e96aa5c735e365cc8bb0e02cef9806713bf6b":{"kind":"REWORK","_number":6,"created":"2021-07-21 10:43:04.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/6"}}},"commit":{"parents":[{"commit":"7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-21 10:42:37.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/173e96aa5c735e365cc8bb0e02cef9806713bf6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/173e96aa5c735e365cc8bb0e02cef9806713bf6b"}]},"branch":"refs/heads/master"},"48884ca3b8d0b97ab6c26add1b8d90bc0ed0df7c":{"kind":"REWORK","_number":7,"created":"2021-07-23 11:28:06.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/7"}}},"commit":{"parents":[{"commit":"7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-23 11:27:34.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48884ca3b8d0b97ab6c26add1b8d90bc0ed0df7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48884ca3b8d0b97ab6c26add1b8d90bc0ed0df7c"}]},"branch":"refs/heads/master"},"56aa88ba9bc02925f75f4288890d08f3ef17398f":{"kind":"REWORK","_number":8,"created":"2021-07-23 11:33:16.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/8"}}},"commit":{"parents":[{"commit":"7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-23 11:32:50.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/56aa88ba9bc02925f75f4288890d08f3ef17398f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/56aa88ba9bc02925f75f4288890d08f3ef17398f"}]},"branch":"refs/heads/master"},"a02e9bf80386e3383acbe9c04241afc1f1d0902a":{"kind":"REWORK","_number":9,"created":"2021-07-26 11:01:03.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/9"}}},"commit":{"parents":[{"commit":"7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-26 11:00:36.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a02e9bf80386e3383acbe9c04241afc1f1d0902a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a02e9bf80386e3383acbe9c04241afc1f1d0902a"}]},"branch":"refs/heads/master"},"25052f439b8d2004b181f628575e5e81c38b7250":{"kind":"REWORK","_number":10,"created":"2021-07-27 10:23:10.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/10"}}},"commit":{"parents":[{"commit":"6ff9425bd2117a19d8e3c5c70e404edb1b736874","subject":"Merge \"Replace deprecated inspect.getargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ff9425bd2117a19d8e3c5c70e404edb1b736874"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-27 10:22:48.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/25052f439b8d2004b181f628575e5e81c38b7250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/25052f439b8d2004b181f628575e5e81c38b7250"}]},"branch":"refs/heads/master"},"984e9275386b856b17ae2669cc145b5f4a6cc183":{"kind":"REWORK","_number":11,"created":"2021-07-27 13:03:18.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/11"}}},"commit":{"parents":[{"commit":"6ff9425bd2117a19d8e3c5c70e404edb1b736874","subject":"Merge \"Replace deprecated inspect.getargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ff9425bd2117a19d8e3c5c70e404edb1b736874"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-27 13:02:32.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/984e9275386b856b17ae2669cc145b5f4a6cc183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/984e9275386b856b17ae2669cc145b5f4a6cc183"}]},"branch":"refs/heads/master"},"417e277bab67bd6485ddafbc70c0699de9894985":{"kind":"REWORK","_number":12,"created":"2021-07-27 13:06:34.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/12"}}},"commit":{"parents":[{"commit":"6ff9425bd2117a19d8e3c5c70e404edb1b736874","subject":"Merge \"Replace deprecated inspect.getargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ff9425bd2117a19d8e3c5c70e404edb1b736874"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-27 13:06:21.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/417e277bab67bd6485ddafbc70c0699de9894985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/417e277bab67bd6485ddafbc70c0699de9894985"}]},"branch":"refs/heads/master"},"2b46c1f74cbf076fca1a100a978a3a8a23d0338c":{"kind":"REWORK","_number":13,"created":"2021-07-27 13:11:48.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/23/747523/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/747523/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/747523/13"}}},"commit":{"parents":[{"commit":"6ff9425bd2117a19d8e3c5c70e404edb1b736874","subject":"Merge \"Replace deprecated inspect.getargspec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6ff9425bd2117a19d8e3c5c70e404edb1b736874"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-08-22 09:26:12.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2021-07-27 13:11:30.000000000","tz":-240},"subject":"Adds l3-ndp-proxy extension api definition","message":"Adds l3-ndp-proxy extension api definition\n\nChange-Id: I26f7e642214bd09a6d4608c9cfdf8c5ef2537489\nPartial-Bug: #1877301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b46c1f74cbf076fca1a100a978a3a8a23d0338c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b46c1f74cbf076fca1a100a978a3a8a23d0338c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
