)]}'
{"id":"openstack%2Fneutron-lib~748030","triplet_id":"openstack%2Fneutron-lib~master~I6a3cd961d6597c95dcd89e4263bd7894a70347ef","project":"openstack/neutron-lib","branch":"master","topic":"bp/address-groups-in-sg-rules","hashtags":[],"change_id":"I6a3cd961d6597c95dcd89e4263bd7894a70347ef","subject":"Add remote_address_group_id field in SG rules","status":"MERGED","created":"2020-08-25 21:08:20.000000000","updated":"2020-09-02 11:23:30.000000000","submitted":"2020-09-02 11:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748030-1599045691115-1545f216","meta_rev_id":"ca5f30878dc3f85436af719539a1a6a45e29958c","_number":748030,"virtual_id_number":748030,"owner":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 11:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-08-31 12:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-08-31 12:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-08-31 12:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 22:46:57.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-08-31 12:28:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-08-31 12:40:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-02 11:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebebc7f2931201c31d6abca885376c46c637cb8b","author":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"date":"2020-08-25 21:08:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd9400987dc0613016e319ff1bc322f85636fc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7646efd171794969b4bce6e0c3df6cea : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0baa272102284c618a67afb3e0074a3b : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/471be37a6b8243db8c1806d3c6156d9d : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa2fae9650740d7b2fed17062f74109 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d54e10f9a4564a6e9afeb92ad5d70722 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bae9b5a61a174736b25280c946fd14ad : SUCCESS in 5m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3644480863f54ee49b09605c167a648e : SUCCESS in 1h 16m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bcd280c318b409489bea0ca29da7fda : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d9538431f9936dabfdb9c82840e030c69bff8c44","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-08-25 22:46:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad325e6dc1b86e5757c285f0ec514e22c1015a14","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-28 07:50:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nI left a nit but I am sure that it can work without that constant","accounts_in_message":[],"_revision_number":1},{"id":"0a7e1df1cc70c7a6c816c66e87540c16b4cec51e","author":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"date":"2020-08-28 19:22:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b032f84d3b7d3a1d3531ef8cef16b81c0812c220","author":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"date":"2020-08-28 19:26:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06c1d2117bc587598fa4aee66e558eb363807833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:26:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1859fbec81445d974b9b038c317ad6 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54361fcd9b9d467897a24e0ad0d70422 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf848519979140ed84bccf848810851a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccb319cea3b4394899150dfe4f72e7c : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad317ee7101b4549a9d48d4b1bdcc355 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5edd8ec1b66421ba8a85515c82cc5e9 : SUCCESS in 5m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cce3226f9e7749cdb915fd669b6dfc08 : SUCCESS in 56m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93d16df03b5d4a19a6a9fd670eda8b82 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7a284b7c470847f396af23bd3491bb57 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1d428925dc33e1965b3ec17efedeacdb098afd2f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-31 12:28:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d359c519881d5f26f7cfb796f56a748039f0e3a5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-31 12:40:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75e9c0fa1329dfb701def8dbb0bc6736592039ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:40:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"367b91ab458a93ad32e9f0fc58216afddaa83b19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:22:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c41106edde47bbb486c33da7ffb4e4 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6988268bcf294dcd9d8ebc5ccdcc6bdd : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d96600e10648358185c959d20bcd37 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b639f3a90f4d4cfe88582aab6aff7cae : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1719566f6ed4701a06d34a646fa799b : SUCCESS in 5m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e14acc1b6524fad9334ad8dcb32a344 : FAILURE in 17m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f98f85a58f7409e9215adfc51b2bcde : SUCCESS in 3m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f9f08cfda358458f8fb544221bc1e228 : FAILURE in 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5f5e9bca384927bfe306a8ed1117170b9a9f7953","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-02 08:39:48.000000000","message":"Patch Set 2:\n\nrecheck\nnova-manage aka setuptools50","accounts_in_message":[],"_revision_number":2},{"id":"6ba7d1bf36fff111353d961d1e3927ac646ee6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:52:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ac3016ff384728aa3a29a086ceb414 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c45e8c54a4042728529fc7d728e988f : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caaffab3bf854875a28e70cff768dce6 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a3d1365ebd44ca28f32bb9d2d1f2ddb : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a7554c04af4ac3a01f86bdb4904c01 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea1ca36791f439699a2c142200d6b4b : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e732a45e5a1248ada1cd54ffc582d8c2 : SUCCESS in 53m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/453d7d7735934b918727660c25e38296 : SUCCESS in 4m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bbf5c7ae532240f9a9f95e1a4db60d94 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f11e4fbc055a1ca839808b3257af0060df93f421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 09:53:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"73ba850285725f1481f3edd4a60f041d91178d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:21:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83c7a8dd98b48c39cb3965a7aad5cb8 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51fe711cf5134faeb8aa631e1e3dfba0 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79049ea088ee4d8185c07b890eaac305 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9ee7daab2c9464cbc130e186251b285 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e691106334d942f5bdeca60f6a5de879 : SUCCESS in 4m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cbf1cf1abb6455eb9443e5315e35fcb : SUCCESS in 1h 24m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b6c43b10acd44fab4fe7a1c2d867909 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eecf64aa5eb64c06b243b9af0f2211f2 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1b32129e29479c5e1aaa122127cf59025859f2d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:21:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f7268bc79c02cee25bf37a3e1b341be0e3553e9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:23:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b7763f26134b06b0d5bcf6265e4ef6 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef4d32c2fd6f41b2adfdd1220e80a06a : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d057aa14af05ebeb3d1ab5d636b1150a01a24fa2","revisions":{"642812c4ec0b7d31f66fade51506d3781dff3c8b":{"kind":"REWORK","_number":1,"created":"2020-08-25 21:08:20.000000000","uploader":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"ref":"refs/changes/30/748030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/30/748030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/1"}}},"commit":{"parents":[{"commit":"fe06e46f7ec3b3285c2754b8f00d69657ef32a29","subject":"Merge \"Log time of the apply_func only if it took more than 0.1 second\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe06e46f7ec3b3285c2754b8f00d69657ef32a29"}]}],"author":{"name":"Hang Yang","email":"hangyang@verizonmedia.com","date":"2020-08-25 21:02:40.000000000","tz":-300},"committer":{"name":"Hang Yang","email":"hangyang@verizonmedia.com","date":"2020-08-25 21:07:48.000000000","tz":-300},"subject":"Add remote_address_group field in SG rules","message":"Add remote_address_group field in SG rules\n\nAdd a new field remote_address_group in security group rule\u0027s API definition.\nThis allows users to easily manage SG rules for aggregated IP address blocks.\n\nChange-Id: I6a3cd961d6597c95dcd89e4263bd7894a70347ef\nImplements: blueprint address-groups-in-sg-rules\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/642812c4ec0b7d31f66fade51506d3781dff3c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/642812c4ec0b7d31f66fade51506d3781dff3c8b"}]},"branch":"refs/heads/master"},"d057aa14af05ebeb3d1ab5d636b1150a01a24fa2":{"kind":"REWORK","_number":2,"created":"2020-08-28 19:22:49.000000000","uploader":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"ref":"refs/changes/30/748030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/30/748030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/30/748030/2"}}},"commit":{"parents":[{"commit":"fe06e46f7ec3b3285c2754b8f00d69657ef32a29","subject":"Merge \"Log time of the apply_func only if it took more than 0.1 second\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe06e46f7ec3b3285c2754b8f00d69657ef32a29"}]}],"author":{"name":"Hang Yang","email":"hangyang@verizonmedia.com","date":"2020-08-25 21:02:40.000000000","tz":-300},"committer":{"name":"Hang Yang","email":"hangyang@verizonmedia.com","date":"2020-08-28 19:21:31.000000000","tz":-300},"subject":"Add remote_address_group_id field in SG rules","message":"Add remote_address_group_id field in SG rules\n\nAdd a new field remote_address_group_id in security group rule\u0027s API\ndefinition. This allows users to easily manage SG rules for aggregated IP\naddress blocks.\n\nChange-Id: I6a3cd961d6597c95dcd89e4263bd7894a70347ef\nImplements: blueprint address-groups-in-sg-rules\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d057aa14af05ebeb3d1ab5d636b1150a01a24fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d057aa14af05ebeb3d1ab5d636b1150a01a24fa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
