)]}'
{"id":"openstack%2Fneutron-lib~759166","triplet_id":"openstack%2Fneutron-lib~master~I8171996e1b4d079d398b7d05bc2dc2f35e8ab073","project":"openstack/neutron-lib","branch":"master","topic":"bp/sriov-smartnic-support","hashtags":[],"change_id":"I8171996e1b4d079d398b7d05bc2dc2f35e8ab073","subject":"port-backend API defination","status":"ABANDONED","created":"2020-10-22 02:21:23.000000000","updated":"2020-12-18 13:44:57.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f703735462809e08ae43b8fbb4c1155340819dd4","_number":759166,"virtual_id_number":759166,"owner":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"date":"2020-10-27 16:56:48.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-22 03:22:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"all":[{"value":-1,"date":"2020-10-22 02:22:07.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"date":"2020-10-22 07:04:13.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 03:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-22 07:04:13.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-10-27 16:56:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"ef0003d004df9f5dbb55d7335fa96b6b12900f57","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-10-22 02:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98aa478b45f78ebb5e625a7451fe02ba4ae1b6de","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-10-22 02:22:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f096c27125b5898e773740f9e50a09b5775c9568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 03:22:40.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45f99f789c047538bfae8bdb1ae13a2 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ce5a7c71f3647d2858bf1e6649e3b75 : FAILURE in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90b194179ff84f1f940dceca8e47fb3f : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507231cb41084b22b809659160005829 : FAILURE in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b57282760da44d748c44ae15099cc005 : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4e9d10f5e0c43ea8a6a478df109605b : FAILURE in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/793f02d75363499e9fa730a54cb2f33a : SUCCESS in 59m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d20f405888a245b99ed7e4c1643fefa5 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2e0e9d61b726fc94621ae73dbaff6399ae185fbd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-27 16:56:48.000000000","message":"Patch Set 1:\n\nDo you have a neutron-spec as well for this?\nOr this nova-spec covers neutron changes as well: https://review.opendev.org/742785 ?","accounts_in_message":[],"_revision_number":1},{"id":"a2c4d3404c630cb8a3c829c0df59ebd4f3d01665","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-12-18 13:24:44.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Do you have a neutron-spec as well for this?\n\u003e Or this nova-spec covers neutron changes as well: https://review.opendev.org/742785 ?\n\nThis is covered by [RFE] 1906602 add new extension \"device-profile\" for port.\nAnd, this is rough POC. \n\nRodolfo Alonso got a much prettier implementation: \nhttps://review.opendev.org/c/openstack/neutron-lib/+/767586","accounts_in_message":[],"_revision_number":1},{"id":"f703735462809e08ae43b8fbb4c1155340819dd4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2020-12-18 13:44:57.000000000","message":"Abandoned\n\nrefer to:\nhttps://review.opendev.org/c/openstack/neutron-lib/+/767586","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecfd58bb134096a6e092ef5486b83c7ef37bf7d6","revisions":{"ecfd58bb134096a6e092ef5486b83c7ef37bf7d6":{"kind":"REWORK","_number":1,"created":"2020-10-22 02:21:23.000000000","uploader":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"ref":"refs/changes/66/759166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/759166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/759166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/759166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/759166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/759166/1"}}},"commit":{"parents":[{"commit":"7bacdb553f4206eb8a172db761531eac618552d0","subject":"Add address groups to callbacks resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7bacdb553f4206eb8a172db761531eac618552d0"}]}],"author":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2020-10-15 02:47:18.000000000","tz":0},"committer":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2020-10-22 02:21:28.000000000","tz":480},"subject":"port-backend API defination","message":"port-backend API defination\n\nThis patch adds a new API \"port_backend\". This API aims\nat describing a profile of hardware device which can be\n associate with a neutron port.\n\nImplements: blueprint sriov-smartnic-support\n\nChange-Id: I8171996e1b4d079d398b7d05bc2dc2f35e8ab073\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ecfd58bb134096a6e092ef5486b83c7ef37bf7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ecfd58bb134096a6e092ef5486b83c7ef37bf7d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
