)]}'
{"id":"openstack%2Fneutron-lib~764291","triplet_id":"openstack%2Fneutron-lib~master~I5c5a2431117ec0381b0ad8058863ccb379d943fd","project":"openstack/neutron-lib","branch":"master","topic":"bug/1880532","hashtags":[],"change_id":"I5c5a2431117ec0381b0ad8058863ccb379d943fd","subject":"Add shim extension for ECMP","status":"MERGED","created":"2020-11-26 08:59:38.000000000","updated":"2021-05-12 08:22:58.000000000","submitted":"2021-05-12 08:21:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764291-bug/1880532","meta_rev_id":"c05df2a5e18f40d7cc5f9d58694b6fc6a241de6c","_number":764291,"virtual_id_number":764291,"owner":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 08:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-07 16:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-16 10:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-05-12 06:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 10:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 09:05:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-07 16:09:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"26bc6e17f8bffc464288a7696ca29f540a9f77b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2020-11-26 08:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73006d4124f96063454f332967258b51cecac95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:23:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f7cc0c829e48ec8b4e5094cf5a587a : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19265390c19746c482a9b58bd60483a2 : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aea39f2b565f4590a25e0090eccecaff : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc0ed4dc2564216917e05068a5656dd : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9377e83f1e4766b9ef37f4ac2d7205 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce8c691a7be496f9d322ebb1df90c3e : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/872c2e177c66483ea6c528397004cc5d : SUCCESS in 5m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abe2f7c031274d74846a32aaeaf5b3d8 : SUCCESS in 1h 21m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cff98fc418884d68a8c4a7576ed445a5 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d1d116ba286e45d696e13184d7c9da13 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"334fb7106cc95335b2d20ced081c8494f71125b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2020-11-27 01:47:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4422b9e39ac93cd656502330091297f33fe2bd77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 02:48:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb7f68ca5e148e99a47d2348fad1665 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc31eebed91e4f82be083a6602a9889e : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87238f5ae6504eb6b74d008e0192a1bb : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dc3ff293ce4e8c8c9ac4e9bf6e4a41 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2178e5d0c5a943f9be9c2f756e592fd0 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4dfd02b282e43ffad0f01fc2cd2c1da : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab6134471e3344c5a91678adc2fc312e : SUCCESS in 3m 40s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a50b48a7c8843399c0a8611aa0f4a97 : FAILURE in 56m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b64bbf413d4898a3ab0c4b33797668 : SUCCESS in 2m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a558db8f78c944d39ad65443431f5469 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cdfcd7da0b3bf272fd178fe59d77a3091d19b3ad","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2020-11-27 02:52:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"021640c98942c44a3918042e07b429c886ed07f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 04:49:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44509a0246546a08dcb1f9c2add6255 : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf278c6bf6864c8cbd98b5b922899fcd : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f453491894844eebb307b5a8e3670f3b : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06e8648498a4bd48a19b34e848dc1b9 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f46f1fa6cea49f8a9e07881b3db112b : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d721b0b3862d4b458c58cab73dd89a68 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a556ebc4cab419082a86eba9fb304e5 : SUCCESS in 3m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/215c4ce94046404a80225a4a2c89f538 : FAILURE in 1h 55m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9afa88f6a9d849dbbf983ffe7698d8b8 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb171947ffc34cf29478133672648974 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9c518749c3dc313324c107e54b920a59e992ed32","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2020-11-27 09:18:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f0ca1e9375b7667e81a7591939101dd7a119ec86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 10:53:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe689d78c15e494b80c5f8d485d2c360 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bca0f53b6fa4e19942ae29d6a609325 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2fd2bb2027e4144b57822848af6aed7 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daabd85385884f519cb0b118264b73b9 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a59f77787b194920a496d81699873378 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115cf95a732d4fdc9b57f6314325aea5 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b26ec271edf4a3daf3c88e8eeac7305 : SUCCESS in 4m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16d944a31f0e49a18db539436b44b862 : SUCCESS in 1h 26m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f6fee2ef5147e69e5133257fb7645f : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d2a1dfd1d8134a8aaf723fea04949df1 : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2},{"id":"370b4c5f262ec0a9928d08867f647250f2ad37b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 09:05:50.000000000","message":"Patch Set 2: Code-Review-1\n\n-1 just to hold on with that until spec will be accepted and merged. Except that it\u0027s LGTM for me","accounts_in_message":[],"_revision_number":2},{"id":"8b777f5f3add166cd4b3ebaa1e29a7b515f43b4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2021-03-15 10:39:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14d059adc6711585de0d356bc9f58f7486406d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:17:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be05a895336645b88f607c31f4ca42db : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/636cdb789dc8483eadac4328a202e656 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f10572aa05a4e3ab2b356cb58ba4e61 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d95f083833d437ab8b48ba7f3269459 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0199c36a7ee148528bee74564623ec19 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28c72934593e4479b09ae075c65f32af : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c255f813f3a460caccb4fbbf8e2e74f : SUCCESS in 4m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1cc236dafac4184a39d98b96227f6de : SUCCESS in 1h 31m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8071409930414f418e2d9d477b264fd9 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/838e19f8b2e3462c8fe3ab063884328b : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e3b27c65d7720abe8407d76e10303ad1c57e159c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 10:11:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ad3df6ec89c118278ccc06c35aa47e9e9bd73e24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 16:09:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"697d92568985acc2fa81d58e3499810b56f1c6e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 06:58:45.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"83243413194f4d07a7bc4d85adc0d6229db33f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 06:59:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4fef4cbe961ec01b47c8872ba09f50e451c00d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:21:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2c80837e10443e8fe7cebb6267e20f : SUCCESS in 10m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08135e4ab6e840c5ba4f982788e15d06 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85fd87d1f644bfaad9427ddd6504b6c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/912e9cac79c542ed8352a7648c0d053f : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13fec2f6c7634998b2790348abc5b845 : SUCCESS in 4m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c96a9f72db6e404e98d3409313af0a92 : SUCCESS in 1h 20m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3daba4b44fcb4da2a580ea2194b193fb : SUCCESS in 5m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fdda6c44913e4831b0fa064f4ba963e6 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":3},{"id":"8546cd6f18b2e3f75847c3c9626c4c47dbbaf316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:21:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c05df2a5e18f40d7cc5f9d58694b6fc6a241de6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:22:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f6c02de90d486d804358bf426c8712 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6f37929786475ebf6d25bd01575d60 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"590f3f1d4f65fa61ed601d9ff396f1c5deeee2cf","revisions":{"b9fb713cae9caa9f0b14b116199de35676e30204":{"kind":"REWORK","_number":1,"created":"2020-11-26 08:59:38.000000000","uploader":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"ref":"refs/changes/91/764291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/91/764291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/1"}}},"commit":{"parents":[{"commit":"28373a924b0bc6fab56f5f52067ae607705f6654","subject":"Merge \"Start running pylint for pep8 tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28373a924b0bc6fab56f5f52067ae607705f6654"}]}],"author":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2020-11-26 08:50:07.000000000","tz":480},"committer":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2020-11-26 08:59:33.000000000","tz":480},"subject":"Add shim extension for ECMP","message":"Add shim extension for ECMP\n\nRelated-Bug: #1880532\nChange-Id: I5c5a2431117ec0381b0ad8058863ccb379d943fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9fb713cae9caa9f0b14b116199de35676e30204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b9fb713cae9caa9f0b14b116199de35676e30204"}]},"branch":"refs/heads/master"},"b451a66f1ae93ff356746880ab2bb7d3f0cd2d5d":{"kind":"REWORK","_number":2,"created":"2020-11-27 01:47:12.000000000","uploader":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"ref":"refs/changes/91/764291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/91/764291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/2"}}},"commit":{"parents":[{"commit":"28373a924b0bc6fab56f5f52067ae607705f6654","subject":"Merge \"Start running pylint for pep8 tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28373a924b0bc6fab56f5f52067ae607705f6654"}]}],"author":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2020-11-26 08:50:07.000000000","tz":480},"committer":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2020-11-27 01:46:22.000000000","tz":480},"subject":"Add shim extension for ECMP","message":"Add shim extension for ECMP\n\nRelated-Bug: #1880532\nChange-Id: I5c5a2431117ec0381b0ad8058863ccb379d943fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b451a66f1ae93ff356746880ab2bb7d3f0cd2d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b451a66f1ae93ff356746880ab2bb7d3f0cd2d5d"}]},"branch":"refs/heads/master"},"590f3f1d4f65fa61ed601d9ff396f1c5deeee2cf":{"kind":"REWORK","_number":3,"created":"2021-03-15 10:39:37.000000000","uploader":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"ref":"refs/changes/91/764291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/91/764291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/91/764291/3"}}},"commit":{"parents":[{"commit":"02e070fe099651ad5abea87819c7d3e729885130","subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02e070fe099651ad5abea87819c7d3e729885130"}]}],"author":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2020-11-26 08:50:07.000000000","tz":480},"committer":{"name":"XiaoYu Zhu","email":"z503755743@gmail.com","date":"2021-03-15 10:38:43.000000000","tz":480},"subject":"Add shim extension for ECMP","message":"Add shim extension for ECMP\n\nRelated-Bug: #1880532\nChange-Id: I5c5a2431117ec0381b0ad8058863ccb379d943fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/590f3f1d4f65fa61ed601d9ff396f1c5deeee2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/590f3f1d4f65fa61ed601d9ff396f1c5deeee2cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
