)]}'
{"id":"openstack%2Fneutron-lib~764416","triplet_id":"openstack%2Fneutron-lib~master~I63a41a21784fa6ce6d9c249f41991c0df3346135","project":"openstack/neutron-lib","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I63a41a21784fa6ce6d9c249f41991c0df3346135","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-11-27 06:45:32.000000000","updated":"2020-12-21 19:44:58.000000000","submitted":"2020-12-21 19:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764416-policy-json-to-yaml","meta_rev_id":"f00f57da19cc9de4a5f7200b316bad14545886fd","_number":764416,"virtual_id_number":764416,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-21 19:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-21 16:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-12-21 19:44:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-12-21 19:44:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-12-21 16:52:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 08:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 09:01:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-01 14:35:35.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-12-01 14:39:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"6bfadccc40f0880a6991bdc8c963442604e22324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-27 06:45:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c60a3e0c64c788b2ab5203eb1e8319b9ea3b2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 08:16:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ec4502a6c843b48f8f8ccaf7e76897 : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bd5370273746a882d2e67bfdca5fb4 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0234be5990cf47968f9f5057da450a6c : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/400c8f95f4ac4f32a43c0c15636baf2f : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7c1ef5ccbdd4f1d8f4c8fe4b5adaa31 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0481a932984cfbbaf692b2d5b68360 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efec540812b14bada0f4e79a637d430f : SUCCESS in 5m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e17ecdb5c9a94a9198d685c83fb6c676 : FAILURE in 1h 24m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/203e88abee124e73808f24eb759bb034 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cbfba0165338a842764960ed4e3cc5b151be53e8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-27 20:47:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6174e2d25bd59d2a790e86628474adcacb1a435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 22:29:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10e91bde5ee443c894142aa07a19c8b : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4460844e8e7c41a68819246f11eb6b8f : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7354a0a31d4142a2ae899b5162bebcda : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02fa52709ba04f5e966a6d4a97b94440 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e89a68d70b7f405491cf748933c2d075 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7388d86b542a413ba7a24bb0474be89d : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb4b71df382495caed654a7f4a85768 : SUCCESS in 5m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47d13b2ea4834a43ac14399029db0c8e : SUCCESS in 1h 35m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3f1f20f30d0140459167116e56dd6979 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"91d270e65391c52736fb5cf6c74580a8ee556a66","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-30 09:01:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"407853aa198461c162690cc56ee13deccc792dd7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-01 14:35:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61c059d93718dde189363f160356e4d09eb26705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-01 14:36:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"754f1c95223b6b1fd148ccbaf92b096628bcc719","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-01 14:36:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e181ec4761714ccddd3f15f489418e3ef7239fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:17:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccea52afcd074fa4bd8efb28abceeda2 : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be342b3cfc4463c9038491778d1033c : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a711733eb33348bc853cd225b52e2336 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7fcc9b6ebd444588fa18bcb73e56ad8 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c6d163833c840dabb01fa8b447823ce : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f28312742b14f65a2fbd78e951e844e : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb0b1c466db24c15b86d6a1d94f8843f : SUCCESS in 3m 49s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47d2288b36fc4d3a9a602a0eb2c306ef : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b43915d704ee46c39d0a09fc503dceef : POST_FAILURE in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"767be3e01470c9e000aff28ea6d2c8b074d41572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:18:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccea52afcd074fa4bd8efb28abceeda2 : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be342b3cfc4463c9038491778d1033c : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a711733eb33348bc853cd225b52e2336 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7fcc9b6ebd444588fa18bcb73e56ad8 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c6d163833c840dabb01fa8b447823ce : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f28312742b14f65a2fbd78e951e844e : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb0b1c466db24c15b86d6a1d94f8843f : SUCCESS in 3m 49s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47d2288b36fc4d3a9a602a0eb2c306ef : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b43915d704ee46c39d0a09fc503dceef : POST_FAILURE in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5924c45a672dc44699fadff68a43353db084467b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-01 21:02:15.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fb1342aa851fcf61714ffc9f4f1c368ef024ecf3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-02 22:51:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bdeeb4c01846b0e736834ee56b07ad1f7d163448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:38:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0816e1e87cfe4a3c8eafade1ce12b226 : SUCCESS in 17m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7a05b331774d5090477db6adcfa8ab : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a0ac81204b433abd8aed54dfe2803e : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f31bea52964e049f309694d99adb7c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/358f3051b0b34fa08b45521adec2ebff : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef79d560a154acda594eaadb5df9787 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e63038c07e6e4b9cad25afc57ee06efd : SUCCESS in 3m 54s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d50d22e2eaa34110b590e8c369241cf8 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5e12dbbb4a449c2a593a50a029f4b55 : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5938dc3828a1619adf5e2b04b2fb25d68055018c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-08 08:43:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b19d616fccebdd402c800760f67abb1d61e0232a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 08:43:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a51b74d136f3a08be5d3a19b0ab462df11a4838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 10:07:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ade78fc8274520a4fc26e360927901 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d623ef2f2ada4c99a80fa751e20e0700 : FAILURE in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbc85c7382a42089fe9b5dd470e7df1 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3736151b6b5247a19e38cacc44737d91 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d49293824df2406fa4944aae448d6cd8 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d94ce9acbf1f47c99e1bc8b5e1eb6bcf : SUCCESS in 1h 17m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1c989dc074e14e2aba849f161c464831 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3f2055f5f51b0815cc225847be7cbd260de6049a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 08:31:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a62462714d2c7813c170c1fac0afcc5b2720dd2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:15:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafb6247643e4126861eef2c7ac219f4 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b985e74bf29049c69212591d66db6ea5 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f4d19614acc4bde8326e707123facb5 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd341b56d8c64ceeacbc5778e10258f3 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2135cc3ad9c4d15b88580603f1dac6a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd9d01b788b4ba8a695251edabcb525 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b987b393b8c4b12849c3bd324d2b488 : SUCCESS in 5m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8505183dde843ba89e29d42893256b7 : SUCCESS in 1h 27m 26s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cfc21e0834054b78b3ca36607b878269 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5ebb3e1a6380b7a3d98a40b53c77ee83746ae355","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-11 06:19:32.000000000","message":"Patch Set 2:\n\nFYI: lower-constraints job is failing with Pip 20.3. I am working on a fix of l-c.","accounts_in_message":[],"_revision_number":2},{"id":"21b4f46918e1343abc4a27a07f6b453fe15b2880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 16:52:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c764479d5d8f13c7b22d7324426089b92045045c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 16:52:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d1986e9631200a374c566e57fb86a7db23d3557a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:09:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f944557beed4601902732462031069b : SUCCESS in 12m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b20f38eed624fceb6598bfd4fb80358 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/032521a5b3e945f8ad13f4c49a272609 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f077d11c5feb446bb940101087983568 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c93738740ef49ca804f9d8f5a50b76d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7cca1c72f81471b9a678b241c107b71 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdc083dafed0437c8cf09ea8517134c0 : SUCCESS in 6m 36s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/012afb60f9a44220b3db93985203d7b0 : SUCCESS in 1h 14m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa97d92131a341b9aac13bbb0a241936 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":3},{"id":"676edc292fdf5e512a227738858fe2d230e10b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:09:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e2a8154b7640722d93c8475eb47f483d5841cb08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 19:43:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c67b1f156c49de8a0099e21ca08596 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc4be941a2c042a2bf994cdb3b951859 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85228cd14de340dd87cb2eaacaf6bf45 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b9fb0ca55374e7193ba3ed33ebc6cad : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc1309492fa446f98eb67ab19cd2c7fb : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c16edb8d49b3485fa4b4c7bc2847a806 : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5535063a3a24326b868fa02b459b986 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0d2e985a853f890cc49979cf75fdc2d2b5af661a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 19:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6642a7866102d953ad4b7ae8d237628fa42b724c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 19:44:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4695341ae8d545eb9526efd60a6ceb74 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3cc97ca8060b0b0e0f31691eab8c229e8a08dbd9","revisions":{"738aec52aaa32b99bfcdaf8b68cbf191cb0ec96f":{"kind":"REWORK","_number":1,"created":"2020-11-27 06:45:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/764416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/764416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/1"}}},"commit":{"parents":[{"commit":"28373a924b0bc6fab56f5f52067ae607705f6654","subject":"Merge \"Start running pylint for pep8 tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28373a924b0bc6fab56f5f52067ae607705f6654"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-27 06:43:34.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-27 06:45:31.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso convert the neutron_lib/tests/etc/policy.json and\nneutron_lib/tests/etc/dummy_policy.json to policy.yaml\nfile. Replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I63a41a21784fa6ce6d9c249f41991c0df3346135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/738aec52aaa32b99bfcdaf8b68cbf191cb0ec96f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/738aec52aaa32b99bfcdaf8b68cbf191cb0ec96f"}]},"branch":"refs/heads/master"},"2f7cbf82b065e392c791436d9a9a1bb25c6894ec":{"kind":"REWORK","_number":2,"created":"2020-12-01 14:36:23.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/16/764416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/764416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/2"}}},"commit":{"parents":[{"commit":"28373a924b0bc6fab56f5f52067ae607705f6654","subject":"Merge \"Start running pylint for pep8 tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/28373a924b0bc6fab56f5f52067ae607705f6654"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-27 06:43:34.000000000","tz":-360},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-12-01 14:36:09.000000000","tz":540},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso convert the neutron_lib/tests/etc/policy.json and\nneutron_lib/tests/etc/dummy_policy.json to policy.yaml\nfile. Replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I63a41a21784fa6ce6d9c249f41991c0df3346135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2f7cbf82b065e392c791436d9a9a1bb25c6894ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2f7cbf82b065e392c791436d9a9a1bb25c6894ec"}]},"branch":"refs/heads/master"},"3cc97ca8060b0b0e0f31691eab8c229e8a08dbd9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-21 16:52:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/16/764416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/764416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/764416/3"}}},"commit":{"parents":[{"commit":"6e95fff9cf6ad1980206fdd33f3bd4b86ee15515","subject":"Update lower-constraints for new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6e95fff9cf6ad1980206fdd33f3bd4b86ee15515"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-27 06:43:34.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-21 16:52:50.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso convert the neutron_lib/tests/etc/policy.json and\nneutron_lib/tests/etc/dummy_policy.json to policy.yaml\nfile. Replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I63a41a21784fa6ce6d9c249f41991c0df3346135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cc97ca8060b0b0e0f31691eab8c229e8a08dbd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3cc97ca8060b0b0e0f31691eab8c229e8a08dbd9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
