)]}'
{"id":"openstack%2Fneutron-lib~767586","triplet_id":"openstack%2Fneutron-lib~master~Ia235fb60405cfba04114d6c825f430bf792b7533","project":"openstack/neutron-lib","branch":"master","topic":"bug/1906602","hashtags":[],"change_id":"Ia235fb60405cfba04114d6c825f430bf792b7533","subject":"New api-def: port-device-profile","status":"MERGED","created":"2020-12-17 16:52:39.000000000","updated":"2020-12-23 04:23:38.000000000","submitted":"2020-12-23 04:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767586-bug/1906602","meta_rev_id":"3e8443b29046414aeb2c2c9d8aa70ee4bca9eb39","_number":767586,"virtual_id_number":767586,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 04:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-12-22 09:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-12-23 02:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-12-23 02:53:24.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 18:18:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 09:10:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2020-12-18 09:20:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-12-18 13:35:16.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":15447,"name":"yongfeng","email":"dolpher.du@intel.com","username":"dolpher"},"state":"REVIEWER"},{"updated":"2020-12-18 13:35:16.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"state":"REVIEWER"},{"updated":"2020-12-21 12:31:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-12-21 12:31:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-12-21 12:31:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7ba666b6bcc41d78c620034488cf7c6443e4de45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-17 16:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d31ef04e5a41a58e3e2a09b701ea88d5165a5caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 18:18:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5bb6c1104d9440669077b6f3c45be9f7 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6446d6c33f9a4a0eb507801a55242e7f : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0395b86c82f949c687b232bbe9a2ec71 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6242dcfa8eb45a48561025d0c6b17fe : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b1d0a4887d43138667a7adb893dba4 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03da8bcca969427490a035458b860808 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9580f36aeb3429da6cb978580792c17 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6fe292b0cf42d79ac48bc66187555e : SUCCESS in 5m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26a6f5c35270431abf27a2c1f43fce44 : SUCCESS in 1h 20m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a0652c75c546a4812be33ea37c3281 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/641e1e8dcf5849869b7534355c006808 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":1},{"id":"13fcd647821a5926dd385b3bae9bd1852061d7b5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-18 09:10:53.000000000","message":"Patch Set 1:\n\namotoki mentioned (https://review.opendev.org/c/openstack/neutron-lib/+/764416 ) that He is working on this lower-constraints issue","accounts_in_message":[],"_revision_number":1},{"id":"e20fe9dc359307a5ae77969440090d74a6158ba6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-18 09:20:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nJust nits","accounts_in_message":[],"_revision_number":1},{"id":"5c0c02d23db388649670a59cb19d4064228b8cd1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 09:34:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"67d15dab76b66fb03793e9bc3e5eaff8b5517c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 09:35:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"47f28931cac504a828d526e3a5c2edfd70a10528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:02:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b494382c56fb4461808e3f566e961668 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5641e2124bcf4e62b460125ada694a82 : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44784a4a8b824cae833698449a57ccf8 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/456ed3411da04d3d8709d4571cb3dbf4 : FAILURE in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf9c73477d643b087267055d72f405f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c277c36565284096aefb92fcaba1b6e5 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb606efa121044bd85a335ce25cdecf7 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c660b55f0af44d19973447258acd6fa4 : SUCCESS in 3m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/add859c45c9045ae99bae0fcc7c71744 : SUCCESS in 1h 17m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9d5e656962f4eb28566b4d16f6a3e10 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/705c8234acc34ecca9ee4abf25e7954b : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2f7e184b5e96c690bab796321970583f9e19df07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 12:26:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"91051d9e0b7754555a6fb52dc6d7387119b5d3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:54:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e31987172d0645eebfc5f84023f01fd4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcfefc9930148ecaad53cec2650ff86 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86c65873f8244248a888d2b9db74cac : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77af4919cf3947fd8eb02b6db21198c1 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3184d6c5f0c44ee695a37c4de2b0e7f2 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36bf4ff770cf4f34a218cc0f32643091 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d178a0fcc87247aaba5a632581322699 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2c47c11235f4ae290062c8407d11677 : SUCCESS in 3m 47s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75439dcceb024501a6803328f93a110d : SUCCESS in 1h 23m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c5e672604944bf9e57ad4bb3c4be03 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2aafe44bc132445eb84348da77e4ec0a : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":3},{"id":"22d20032b3627e237505462dea39641c05ea24d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 16:22:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3099ffe0923d3a446c0f9858022950eaf33a958f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 16:43:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71b831419a68a46b12c69fcdef8e04ac70648292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 16:58:39.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c227921a8e9debc70ed3c5a0ee6e37f4f788bc19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:19:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5442e33485e459aad8b493f9a9787cb : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2d6da3728d4f139a3690c4db20f12c : SUCCESS in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc84242e092f4af8aec3b5e1c79061a5 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bf51ad19a784977a50bb8433fa9b836 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5a0699a4974b7798bbc24dbe44d717 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c7bdf1944341aba2683e4faa90db9c : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd1a0a6f53ae42d78e386cd57d976620 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5a0833c5fa443b5a41b8cec4039d89b : SUCCESS in 4m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/088d89c729704d7882b82a69a5afa90d : SUCCESS in 1h 15m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba9c485a494c44788540298f0e5feeae : SUCCESS in 3m 07s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/92a74105baef422da2a6014127085bff : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":6},{"id":"0784a48ac76832277e4331cbf5605dea11f6960e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-22 09:46:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a01cbc9f3e12d2ead7271d9ed3407c324bceb576","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-23 02:53:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6785e5d2525f0c99be367e71ca2a7c44c8ba03e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 02:53:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dacfbcb2b0492f896713f585c4ca02cbc52bee8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 04:22:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b113b1977e7743fa8e0d664ed168d249 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb81bd95afd463da5906c5a04c6714e : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ebaeac1351246e096a5507d44c3edc7 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07dd3a8c528949869b8a31610369847c : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b5491a9749b412bb698af8325876631 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/967ab46795c94e2fa0ba9b564e41339c : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03efc24d82f247618cd567d5bf662765 : SUCCESS in 1h 24m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3350c3df7e1744929b844831b3855fbd : SUCCESS in 3m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef60c9483e6442b18fc401c839f08c83 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":6},{"id":"118f2b426f7e32d96e5b10720beed5572381f574","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 04:22:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"3e8443b29046414aeb2c2c9d8aa70ee4bca9eb39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 04:23:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f39cbf9a260c4bd1b1ed7b1af9a19f1e : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bb0126cc8b477098fed9958ca46be3 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75bcb59dc7db4cb78ea060cabb95530a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a4987c747c2171111a5abca11316832ade07a83b","revisions":{"b08c98fb6ee8c3c4553c411934dcf2454bd6b025":{"kind":"REWORK","_number":1,"created":"2020-12-17 16:52:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/1"}}},"commit":{"parents":[{"commit":"d80dc7f54a92cceb24201e061be2ef80c4ab4266","subject":"Remove \"autonested_transaction\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d80dc7f54a92cceb24201e061be2ef80c4ab4266"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:52:30.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b08c98fb6ee8c3c4553c411934dcf2454bd6b025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b08c98fb6ee8c3c4553c411934dcf2454bd6b025"}]},"branch":"refs/heads/master"},"80ab36d64ef10af9724c3b76fd30c75fefe6b3de":{"kind":"REWORK","_number":2,"created":"2020-12-18 09:35:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/2"}}},"commit":{"parents":[{"commit":"d80dc7f54a92cceb24201e061be2ef80c4ab4266","subject":"Remove \"autonested_transaction\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d80dc7f54a92cceb24201e061be2ef80c4ab4266"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-18 09:35:23.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/80ab36d64ef10af9724c3b76fd30c75fefe6b3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/80ab36d64ef10af9724c3b76fd30c75fefe6b3de"}]},"branch":"refs/heads/master"},"e64e048d8c7690ceba982e89df182b2d3cb87a52":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-18 12:26:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/3"}}},"commit":{"parents":[{"commit":"2649bb736dd9f40654c383814bf48c15c07dd022","subject":"Bump PyYAML to 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2649bb736dd9f40654c383814bf48c15c07dd022"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-18 12:26:10.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e64e048d8c7690ceba982e89df182b2d3cb87a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e64e048d8c7690ceba982e89df182b2d3cb87a52"}]},"branch":"refs/heads/master"},"f9722c98f0102d59d44511967e6c6965b0b1e6be":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-12-18 16:22:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/4"}}},"commit":{"parents":[{"commit":"ae9c6f6f7c293c17d2b8067b44ed0c13c2c87719","subject":"Update lower-constraints for new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ae9c6f6f7c293c17d2b8067b44ed0c13c2c87719"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-18 16:22:19.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f9722c98f0102d59d44511967e6c6965b0b1e6be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f9722c98f0102d59d44511967e6c6965b0b1e6be"}]},"branch":"refs/heads/master"},"64a40007eab7f605b235b785584abb5016270001":{"kind":"REWORK","_number":5,"created":"2020-12-18 16:43:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/5"}}},"commit":{"parents":[{"commit":"ae9c6f6f7c293c17d2b8067b44ed0c13c2c87719","subject":"Update lower-constraints for new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ae9c6f6f7c293c17d2b8067b44ed0c13c2c87719"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-18 16:43:17.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64a40007eab7f605b235b785584abb5016270001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64a40007eab7f605b235b785584abb5016270001"}]},"branch":"refs/heads/master"},"a4987c747c2171111a5abca11316832ade07a83b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-12-18 16:58:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/86/767586/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/86/767586/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/86/767586/6"}}},"commit":{"parents":[{"commit":"6e95fff9cf6ad1980206fdd33f3bd4b86ee15515","subject":"Update lower-constraints for new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6e95fff9cf6ad1980206fdd33f3bd4b86ee15515"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-17 16:50:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-18 16:58:39.000000000","tz":0},"subject":"New api-def: port-device-profile","message":"New api-def: port-device-profile\n\nNew API definition: ``port-device-profile``. The new field added to\n``ports`` API, ``device_profile``, represents a Cyborg device\nprofile and is a named set of user requirements for one or more\naccelerators.\n\nRelated-Bug: #1906602\n\nChange-Id: Ia235fb60405cfba04114d6c825f430bf792b7533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a4987c747c2171111a5abca11316832ade07a83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a4987c747c2171111a5abca11316832ade07a83b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
