)]}'
{"id":"openstack%2Fneutron-lib~768324","triplet_id":"openstack%2Fneutron-lib~master~I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb","project":"openstack/neutron-lib","branch":"master","topic":"bug/1909100","hashtags":[],"change_id":"I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb","subject":"Add new VNIC types for Cyborg provisioned ports","status":"MERGED","created":"2020-12-23 11:03:40.000000000","updated":"2021-01-13 19:52:36.000000000","submitted":"2021-01-13 19:51:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"768324-bug/1909100","meta_rev_id":"99aec14429ceff9252b052e75a55e13a49004971","_number":768324,"virtual_id_number":768324,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 19:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-01-13 01:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":2,"date":"2021-01-13 09:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2021-01-12 10:44:55.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":1,"date":"2021-01-13 09:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-23 12:32:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-11 05:46:23.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2021-01-11 05:46:23.000000000","updated_by":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"reviewer":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"state":"REVIEWER"},{"updated":"2021-01-11 10:20:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-12 06:58:59.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"8c4626f9b713d349fe323db4ef1da13a863ab49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-23 11:03:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cac7d54065b2281103e18dfa7d6f56f04d9662f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-23 11:04:06.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"76cc417514c9838c1cdddc232e28dc17c8c71402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 12:32:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f03b5fc954640038039f3f40322bf14 : SUCCESS in 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/322c78c806f646bca33e6aaf7f6bf806 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f667e040fc848fd9bf0d656d95b8060 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d2d1474572d4bfd9f7b93851afdfc6b : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9137c3132bb5475e8c6475662d1849bf : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d48adbea1c6641b7902d145754aab7ff : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c446361f6d9a4f64a1cab61a50f71ef4 : SUCCESS in 3m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b23ec79370ea4a34a861c8a5f8726cb8 : SUCCESS in 1h 23m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2886e33e82fd4d65a811cda0dfd1032c : SUCCESS in 2m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b9cb237b33243fcbdae1a6ee9e6aecf : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a2db18c155d941407a8900804db6c12756e9334a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-08 15:48:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b07d42f71b8b59547663e8befe4223f86bd0040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-08 15:49:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73e3d6480fecf6d89ccde7919731a878f1211256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:23:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1ed30bf7e142fe95810edfffdb0ad4 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/234ed761ec284bc08f071b3301ab3ae3 : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b727e8d475614540adb49565924763be : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9581834908ce42798b82aa5ac1ac6070 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0293848a5e9f4a09b3adab91d8f440e4 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe13a6b6612b4930a558f46924ad4e51 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f1f8c4af9ef442bbd3301deab343d97 : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/781182cab0244cc79da1811022bfaba7 : SUCCESS in 1h 27m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec1b29df1d654f5381ae41df29b9459f : SUCCESS in 3m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2dfda2adaef7454484f4e8221927f524 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7db3a82fb87273f5ca05e1aa44c28483f29b8c43","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-11 10:20:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5314d2beadc5f9da125f85b37c229fd6c0db914b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-01-11 11:32:25.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for your update, it looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"5d613997e4be282fdf060295b8087204166fbf0a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-12 06:58:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cf9a28ee5a2829b332ebb212ae2c85496e01bf4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 10:11:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a5e19dd9b3e2c79eade5519dd5fe313e6159ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 10:13:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"986b694359fce189aa1bfba783d0bb4829abf090","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-12 10:44:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e58bfec81ecf67a4cd18379cdb3815854c1a4394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:55:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7b030c9b28459587d7fff769a2b1d3 : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85b1267863c846ca8515ccaed87a2a9a : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f262a0b3d394399870ccf73b1eebb7e : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e49a8eeaf4849228a9371ec33924939 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781d337cc74c4b93926f201a29a8e05a : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a4dbdcbadbd4b828da744aa7c0f184c : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14faddad25564a8d86d28a14cf3eea94 : SUCCESS in 9m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e9d562f0fa04562ba46f4f8b79f4a3d : SUCCESS in 1h 35m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eaecfb2da7c42b4a79f82c51d5dfac4 : SUCCESS in 5m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6332501991d4484aad09e188bdfac873 : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":4},{"id":"f8b6be1518266f152e0b953ae11fabcd3108007e","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2021-01-13 01:35:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"24c44c5cccb4b68a6e4d28507ebd329a02ae34b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 09:43:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6c7d25a182cf4c159c8c7b2d20e64e827d96fb50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 09:43:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9cc0502ed073f1207010ca8ac00164edc8523d5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:08:36.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2324219c855747dd873a95e8d0c8cd16 : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d2737b737474655a737e2dec98d285e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed0581cd7e0477ca07434a56da729d3 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a242fbd7824054b00543db4ce6337a : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c6e8e6024d94e20b3944e4e9bbc4606 : SUCCESS in 5m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f732a288aed4caf872350a9c3351cd1 : SUCCESS in 1h 21m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14cc335582df4c319d41b6cf23570835 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/24be740609404e24a01562750c6127e7 : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"efbb0bd334f077809ffd3fae685bdb5c6daf5590","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 11:14:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9886e40cca2afd59c3b273b9995ae4922016927c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:28:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc3d920a80b49e6abdc45c7b637b0ec : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8c826efb5a34661b648910b87780f2f : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/817289cca77f485d8196cf46332fe7de : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21842970267c42899cf597d8439dcf39 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564cad22c5264db5845a7e72afb2230d : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb7e5d2ac78646e284e73a8640e859c5 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7475a76c557e4a4694e0ae736d5fde63 : SUCCESS in 4m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/caeedb28097b4880bc5182e89a07f069 : FAILURE in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b47d8cbb2be4e57b622964dc12f85eb : SUCCESS in 6m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c48a9a4d52cd49158b4b38507beecd50 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0b7ceafc8805af71bbdda877e1a9f426bf34302b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 16:30:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6d6d81984bd367a816ad833cc26a3139cd267de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:08:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb44129fe6146c3b71e2bfcaa322194 : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7bd45cea2d248c4b9789ecfb9285e5b : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5d680a5375a4f16b3478db0abe37dd9 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49399da62f0d4d2f8e831787a60c5320 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1b280f70d394392a2786e6836f1d224 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ba2ea9b130b4e9695c92cbc7cc14bb0 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12357b07dae348a1b317cc111ba8c8e7 : SUCCESS in 4m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa86a7ea5eaf4a35bd6ab13df06426b9 : SUCCESS in 1h 25m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a368aebaace441585579c40fb6e4570 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b6c47a204cc4301b7aa667cd9cda4e0 : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":4},{"id":"43fd95748c953b5dffb4d5f77fdada413359112d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:09:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"81b1e124f2521dfa49b0d82ea46979303d86d6f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:51:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f795f347c39142a5b5f1cf9dbda342b5 : SUCCESS in 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9771f2d6ccb748f8b45c9e805144d483 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfb9d1e02854233a1c9aab07a136bc8 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9363d9bd0744f849fc64eee54c02764 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf54e79ac244c1a9b102a0537969417 : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89b64e8ff4e346158a2db7c10e79d0f9 : SUCCESS in 1h 34m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd354451ec8440883d681fbd2ab325a : SUCCESS in 4m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2d3749391eb24ce0851683e77190783b : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7dc32d5ebc886d9eca322d4646a1031157ccd613","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:51:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"99aec14429ceff9252b052e75a55e13a49004971","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:52:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e51745e9fd4a40a07a086d040afb04 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4dc303b51a34d639bdd70b0858fb95c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"061571770043dca89004c3659578415110037f74","revisions":{"397e51f8e211bcc5a78188e8cbb373c13b8c8728":{"kind":"REWORK","_number":1,"created":"2020-12-23 11:03:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/24/768324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/768324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/1"}}},"commit":{"parents":[{"commit":"c0cf1b35aad76d648f42720a23871f8cbd6c6887","subject":"Merge \"New api-def: port-device-profile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c0cf1b35aad76d648f42720a23871f8cbd6c6887"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 11:02:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 11:02:54.000000000","tz":0},"subject":"Add new VNIC type ``cyborg``","message":"Add new VNIC type ``cyborg``\n\nRepresents a port that supports any kind of hardware acceleration\nand is provided by Cyborg\n\nChange-Id: I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb\nRelated-Bug: #1909100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/397e51f8e211bcc5a78188e8cbb373c13b8c8728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/397e51f8e211bcc5a78188e8cbb373c13b8c8728"}]},"branch":"refs/heads/master"},"cdbe387c20135581ef0dc6faff583c03667d4a5d":{"kind":"REWORK","_number":2,"created":"2021-01-08 15:48:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/24/768324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/768324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/2"}}},"commit":{"parents":[{"commit":"307b6be55e233af3c943b137a93a43cf3f899ab3","subject":"Merge \"Add `access_as_readonly` sharing action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/307b6be55e233af3c943b137a93a43cf3f899ab3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 11:02:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-08 15:47:27.000000000","tz":0},"subject":"Add new VNIC types for Cyborg provisioned ports","message":"Add new VNIC types for Cyborg provisioned ports\n\nAdded VNIC types ``accelerator-direct`` and\n``accelerator-direct-physical``. Represents a port that supports any\nkind of hardware acceleration and is provided by Cyborg\n\nChange-Id: I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb\nRelated-Bug: #1909100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cdbe387c20135581ef0dc6faff583c03667d4a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cdbe387c20135581ef0dc6faff583c03667d4a5d"}]},"branch":"refs/heads/master"},"4be9f95f842bc9ce47ceac33cf545636bde80573":{"kind":"REWORK","_number":3,"created":"2021-01-08 15:49:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/24/768324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/768324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/3"}}},"commit":{"parents":[{"commit":"307b6be55e233af3c943b137a93a43cf3f899ab3","subject":"Merge \"Add `access_as_readonly` sharing action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/307b6be55e233af3c943b137a93a43cf3f899ab3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 11:02:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-08 15:49:03.000000000","tz":0},"subject":"Add new VNIC types for Cyborg provisioned ports","message":"Add new VNIC types for Cyborg provisioned ports\n\nAdded VNIC types ``accelerator-direct`` and\n``accelerator-direct-physical``. Represents a port that supports any\nkind of hardware acceleration and is provided by Cyborg\n\nChange-Id: I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb\nRelated-Bug: #1909100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4be9f95f842bc9ce47ceac33cf545636bde80573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4be9f95f842bc9ce47ceac33cf545636bde80573"}]},"branch":"refs/heads/master"},"061571770043dca89004c3659578415110037f74":{"kind":"REWORK","_number":4,"created":"2021-01-12 10:13:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/24/768324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/768324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/768324/4"}}},"commit":{"parents":[{"commit":"307b6be55e233af3c943b137a93a43cf3f899ab3","subject":"Merge \"Add `access_as_readonly` sharing action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/307b6be55e233af3c943b137a93a43cf3f899ab3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 11:02:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 10:13:02.000000000","tz":0},"subject":"Add new VNIC types for Cyborg provisioned ports","message":"Add new VNIC types for Cyborg provisioned ports\n\nAdded VNIC types ``accelerator-direct`` and\n``accelerator-direct-physical``. Represents a port that supports any\nkind of hardware acceleration and is provided by Cyborg\n\nChange-Id: I32e16a7cf3809e2469b3e7a7e068c8f0849ccdfb\nRelated-Bug: #1909100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/061571770043dca89004c3659578415110037f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/061571770043dca89004c3659578415110037f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
