)]}'
{"id":"openstack%2Fneutron-lib~770612","triplet_id":"openstack%2Fneutron-lib~master~I7a674dacab343b89f101f83a78243ec971ce9712","project":"openstack/neutron-lib","branch":"master","topic":"bug/1906318","hashtags":[],"change_id":"I7a674dacab343b89f101f83a78243ec971ce9712","subject":"Add network parameter to some of the ML2 API methods","status":"MERGED","created":"2021-01-13 13:06:41.000000000","updated":"2021-06-10 06:17:30.000000000","submitted":"2021-04-23 10:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770612-bug/1906318","meta_rev_id":"bccc05e7b0be5ad572145fceff6ba784b2a6baac","_number":770612,"virtual_id_number":770612,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 10:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-02 14:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-01 16:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-04-23 09:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2021-03-02 07:26:14.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-13 13:07:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-01-13 14:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8da41c7f1cb4ef48f65c2b22c04175aaf2c5994e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 13:06:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35d92b6f31fec06085116b8ac86f157d47ab37b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:44:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7e9b6a23bf4033bff222ad08f329cd : SUCCESS in 15m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa5a2b0c08274861b66d048d67c521ad : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa1bbe79af1e443bbe9dc0721f6b770e : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea382b5b90847249d1e7034cead271a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04cbf8c60d34a24a2b8d26b52434b81 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76d69a7fdf0c4f8690ed91c4880530af : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8133c355b58c4081a5509f2c931af2e4 : SUCCESS in 6m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68d50d421c274fa1a2206ce781bc42a2 : SUCCESS in 1h 33m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a6b6fd897d0e47a78be06121647167dd : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f77179ea14266401602caabb3829572d9d0a4bd8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-13 17:04:57.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nRelease note too?\n\nAnd I\u0027m guess I\u0027m fine with this as it\u0027s an optional argument.","accounts_in_message":[],"_revision_number":1},{"id":"780538b165ccf41df887753b8c38d9df5c338a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-14 11:55:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95f5be33a710a11c67e62e5fa2ff3c7da2a778ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:24:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46832903232248c286f8ca309015737d : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a83b7ae4cf8b4613b33804fbcff0b869 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0e3e7aa85141b6ab3145b6e753a3a3 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8e6eb6a1dae4b33ac7d2d39c0a50c14 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ba3b34db828402496840df071210cd4 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c7479afb2ce4d6e8958852794983e00 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bcc5a4ed0b4aae87d65bbab4eb0fef : SUCCESS in 4m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58b857ae22bb4892a15d6f40bc8e67a6 : SUCCESS in 1h 19m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d736d618e7c4a6384a0c97665c8734d : SUCCESS in 4m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/95d641c2ddfe4c229003bab7f513f6c9 : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0cdff9c8764e5310d5f98a24c0e2dea109794204","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-05 15:58:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e82503a48f6faa82da16e06ad7725a62e0846ca","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-01 09:42:14.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"3464e04d5fd724e638b91d818bd55d638d8f983d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-01 16:52:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"160e5174883e7ad261af5b96909a1a0bf088d760","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 07:26:14.000000000","message":"Patch Set 2: -Review-Priority","accounts_in_message":[],"_revision_number":2},{"id":"6660eae2ef74a70a558dd68c7a6207312d5119b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-02 14:46:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c5b05d529cbe6296e62bdafeb921dca6ab0931ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-23 09:08:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9cb143482a976f83188af5fe37f40dc9259ed9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:08:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd5651b6d389521264b12b0e676360146924a4e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 10:33:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7206b9960b41c6b89e8cef95694459 : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5502e10f1d6d4f4e9b58946773648e25 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9d5d127d1864db8b9d4831fd7ec4400 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4068c368f7a4fcc9e95ec6a37222e25 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e02c7153ff84d6cabd3dab13b77cb1a : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de96b506c4f346ba9d416cc49e00a03d : SUCCESS in 1h 23m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35f8f4189abb491eb405c8f54373fc40 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d8d774e7910c4c0fa32ecd43bc95efee : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8465e03ccaccd92e263a7c28fd64097fc335d204","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 10:33:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fe93018de8d28407bf0dded5365edad2f98d76e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 10:35:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c018e04304f34ac895d9557aa8969992 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b744d1a491144d88d646c81d22abab4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bccc05e7b0be5ad572145fceff6ba784b2a6baac","tag":"autogenerated:gerrit:revert","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-10 06:17:30.000000000","message":"Created a revert of this change as I01cb252c423cac6a529a785d5dce7f6b3cbe3b5d","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5adf98e857a2ddb923ba0d47a042027122962a5a","revisions":{"2b355abf18c99b243940dbe2ec9700ac88a55c5d":{"kind":"REWORK","_number":1,"created":"2021-01-13 13:06:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/770612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/770612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/1"}}},"commit":{"parents":[{"commit":"307b6be55e233af3c943b137a93a43cf3f899ab3","subject":"Merge \"Add `access_as_readonly` sharing action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/307b6be55e233af3c943b137a93a43cf3f899ab3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-13 12:59:52.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-13 13:06:37.000000000","tz":60},"subject":"Add network parameter to some of the ML2 API methods","message":"Add network parameter to some of the ML2 API methods\n\nFor networks with enabled vlan transparency, get_mtu() method of the\ntype drivers should return MTU value lowered by 4 bytes from the\nstandard value. Those 4 bytes are needed for potentially additional vlan\nheader which user will add if vlan interface will be configured in such\nnetwork on the instances.\n\nTo be able to include that, get_mtu() method needs to know if network\nhas enabled vlan_transparency.\nThis patch adds optional \"network\" attribute to be passed to the\nget_mtu() method but also to reserve_provider_segment() and\nallocate_tenant_segment() methods of the base type driver classes.\n\nMethods reserve_provider_segment() and allocate_tenant_segment() needs\nto have this network dict passed also as those methods are calling\nget_mtu() method in such type drivers like e.g. vlan or flat.\n\nRelated-bug: #1906318\nChange-Id: I7a674dacab343b89f101f83a78243ec971ce9712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b355abf18c99b243940dbe2ec9700ac88a55c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b355abf18c99b243940dbe2ec9700ac88a55c5d"}]},"branch":"refs/heads/master"},"5adf98e857a2ddb923ba0d47a042027122962a5a":{"kind":"REWORK","_number":2,"created":"2021-01-14 11:55:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/770612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/12/770612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/12/770612/2"}}},"commit":{"parents":[{"commit":"307b6be55e233af3c943b137a93a43cf3f899ab3","subject":"Merge \"Add `access_as_readonly` sharing action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/307b6be55e233af3c943b137a93a43cf3f899ab3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-13 12:59:52.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-14 11:55:32.000000000","tz":60},"subject":"Add network parameter to some of the ML2 API methods","message":"Add network parameter to some of the ML2 API methods\n\nFor networks with enabled vlan transparency, get_mtu() method of the\ntype drivers should return MTU value lowered by 4 bytes from the\nstandard value. Those 4 bytes are needed for potentially additional vlan\nheader which user will add if vlan interface will be configured in such\nnetwork on the instances.\n\nTo be able to include that, get_mtu() method needs to know if network\nhas enabled vlan_transparency.\nThis patch adds optional \"network\" attribute to be passed to the\nget_mtu() method but also to reserve_provider_segment() and\nallocate_tenant_segment() methods of the base type driver classes.\n\nMethods reserve_provider_segment() and allocate_tenant_segment() needs\nto have this network dict passed also as those methods are calling\nget_mtu() method in such type drivers like e.g. vlan or flat.\n\nRelated-bug: #1906318\nChange-Id: I7a674dacab343b89f101f83a78243ec971ce9712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5adf98e857a2ddb923ba0d47a042027122962a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5adf98e857a2ddb923ba0d47a042027122962a5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
