)]}'
{"id":"openstack%2Fneutron-lib~771892","triplet_id":"openstack%2Fneutron-lib~master~I530dc3068197d24c9a581a3cd07020ec23132a52","project":"openstack/neutron-lib","branch":"master","topic":"bp/address-groups-in-sg-rules","hashtags":[],"change_id":"I530dc3068197d24c9a581a3cd07020ec23132a52","subject":"Introduce rbac-address-group api extension","status":"MERGED","created":"2021-01-21 22:17:53.000000000","updated":"2021-02-03 17:52:02.000000000","submitted":"2021-02-03 17:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771892-bp/address-groups-in-sg-rules","meta_rev_id":"dd27120e8e97d06184d4b86c19ed1bb38a2c9144","_number":771892,"virtual_id_number":771892,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 17:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-03 16:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-01-29 09:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-26 16:26:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-02-03 16:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 00:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 16:26:45.000000000","updated_by":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"reviewer":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"state":"REVIEWER"},{"updated":"2021-01-29 09:01:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-03 16:12:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"8f51537349e24b719ef8565130ba80798acb272f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-21 22:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bcc290b1a991c11d449432c963e343e45f34b02","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-21 23:18:09.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":1},{"id":"5db4434caca115a1bb3634c5db712f9fe61b8c0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-21 23:18:28.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":1},{"id":"d5787fc6b113b444503a36dc8d9ccd3dae85c13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 00:00:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/17d973dbf1a740bf9525b63586f4f0d2 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a6bbcff8f24d4cbc641274a429c0cc : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d90e1504db5430a95b5426400eb8b9d : FAILURE in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83a95ac1d2044c2f9340d09af4159429 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff23c77069524374a5b2ef0ef1922517 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/888cca8a60d9423ea741a07db99baf4d : FAILURE in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0dbab3c6958479a955565b66919fff4 : FAILURE in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c9b2459f9cd48459703c3ebf5198777 : FAILURE in 5m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04c94d4baadc4f3f84c46a15a3939d23 : SUCCESS in 1h 34m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1836d09b555d4a4497bb598247f2d8a0 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e581cd9aefa444848a223fd5a6b66b2f : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c9c805e236d84a658d69aab1a2735f560f31c55b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-22 00:32:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dffc381855f8fc18e7e33aee79608eba2dfcf219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:42:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d7b9dc2078e4bb4a74fb2a8dcc20b29 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331dacbfc9b6490faa383b9b2bbdadf6 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6933d61b7754865b223e32f90d808fb : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcb46cb77ec34011aeef86ca73b1dff4 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e55e862ece24782b708e31d87dd63b8 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0802c2215f4c20991d84c3d155287d : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bc90b3bcee64bec90670525421c7eb5 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd1b0845b9c44b4d92cac1f0756229ba : SUCCESS in 3m 51s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2093c388f2f74864a593ee6553126bce : SUCCESS in 2h 02m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/485bbd11ba1c46f1b8e68c389cd6f489 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc2135fdd4ed4b27a5f5f0d2a848921f : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":2},{"id":"da9cccd493f5f795624f3a8482786756837115f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:11:07.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":2},{"id":"c4fdac11130bac15307bb1d84c5b30b4a5311f03","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:11:16.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":2},{"id":"26c39c9b70639dd56f34009275be6d83b993b08f","author":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"date":"2021-01-26 16:26:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfdb49ed0f051dbfab9b5189c5c6fc10e95b0aa7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-29 09:01:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfdd398e743977c04facdf4f086e80749ff0650b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-03 16:12:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11fa95a6130088afaedf4a51f4663963d064ba5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:12:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf9c673ddfcf4ba809670b7f3461834d09fa4989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:49:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/415b4ad637c2446289aa4111af46b379 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2ce530c171441fb795204737296e28 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f39141ab3aa94d639149fb760a0e1dd7 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3a618d208a4fbaaa7933f4594ab98a : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f7e45040ef4b8b889f6a8b2fec1dbf : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a79e352406b4493840e405a37df625c : SUCCESS in 4m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7598d298afcf450dbfdb4aa214f99bbb : SUCCESS in 1h 30m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66733cd558484413867e6d6318db8416 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b013b46ac0f74dc291568a3aeef5dce2 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":2},{"id":"54075ede41c647c14a412b69396917e6d497886e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:49:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd27120e8e97d06184d4b86c19ed1bb38a2c9144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:52:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6033fe7120564fb78d89facd9fbf6b27 : SUCCESS in 1m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe4504144414d4e87e061a43ba98a1e : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b62172d979146ed9ce52a85759b8166 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4632e29a41ce08c485d5724971dce9ac55c6a7d7","revisions":{"3a88544f63c4708942f7a83d7bfa0d14f744dad0":{"kind":"REWORK","_number":1,"created":"2021-01-21 22:17:53.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/92/771892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/92/771892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/1"}}},"commit":{"parents":[{"commit":"8c626fe039a5c32e26710b8e8b6d53a4ac6a11b7","subject":"Remove \"tag\" and \"tag-ext\" extensions from \"tags\" documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c626fe039a5c32e26710b8e8b6d53a4ac6a11b7"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-21 22:17:39.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-21 22:17:39.000000000","tz":-360},"subject":"Introduce rbac-address-group api extension","message":"Introduce rbac-address-group api extension\n\nThis extension makes it possible to add address groups to RBAC policies.\n\nChange-Id: I530dc3068197d24c9a581a3cd07020ec23132a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a88544f63c4708942f7a83d7bfa0d14f744dad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a88544f63c4708942f7a83d7bfa0d14f744dad0"}]},"branch":"refs/heads/master"},"4632e29a41ce08c485d5724971dce9ac55c6a7d7":{"kind":"REWORK","_number":2,"created":"2021-01-22 00:32:20.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/92/771892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/92/771892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/92/771892/2"}}},"commit":{"parents":[{"commit":"8c626fe039a5c32e26710b8e8b6d53a4ac6a11b7","subject":"Remove \"tag\" and \"tag-ext\" extensions from \"tags\" documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c626fe039a5c32e26710b8e8b6d53a4ac6a11b7"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-21 22:17:39.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-22 00:32:06.000000000","tz":-360},"subject":"Introduce rbac-address-group api extension","message":"Introduce rbac-address-group api extension\n\nThis extension makes it possible to add address groups to RBAC policies.\n\nChange-Id: I530dc3068197d24c9a581a3cd07020ec23132a52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4632e29a41ce08c485d5724971dce9ac55c6a7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4632e29a41ce08c485d5724971dce9ac55c6a7d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
