)]}'
{"id":"openstack%2Fneutron-lib~774083","triplet_id":"openstack%2Fneutron-lib~master~I6b57f7ce6d411be388c63b87f2c75f2de703cc97","project":"openstack/neutron-lib","branch":"master","topic":"bug/1913180","hashtags":[],"change_id":"I6b57f7ce6d411be388c63b87f2c75f2de703cc97","subject":"Add qos rule validation for network","status":"MERGED","created":"2021-02-04 13:27:35.000000000","updated":"2021-02-05 09:27:24.000000000","submitted":"2021-02-05 09:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"774083-bug/1913180","meta_rev_id":"e646e8136a27450bbbcdbe01deccd14c6f0e04a2","_number":774083,"virtual_id_number":774083,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 09:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-05 08:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-04 14:21:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-02-05 08:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 13:57:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-04 14:07:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-02-04 16:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 08:16:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5331b0a4a369013820e978a5c9fd3f8576717dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-02-04 13:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"803458a5271a6ef8149728f6f41f74cb449eab16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-02-04 13:29:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea0e705774ba2ee7e98b6b386406558bc0277072","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-04 13:57:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8bfddbd2c08b5044a37780277e28ff75d9f335b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-04 14:07:13.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nJust the small nit related to the exception name.","accounts_in_message":[],"_revision_number":2},{"id":"1c5c7e39d4801a09093ef791f4f1f7c766bb2aae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-04 14:08:05.000000000","message":"Patch Set 2:\n\nBTW, push the patch with topic \"bug/1913180\"","accounts_in_message":[],"_revision_number":2},{"id":"6a5ffc2685875a4332850029f5203e040e4ed41e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-02-04 14:16:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b591e5baeb73d3f9f0044fc691c0eabde0eb747","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-02-04 14:19:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b1a7e8128beac5a0805985a37b378d646ddb54c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-04 14:20:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e690d64f740cb7495e18e2706ee384732780356","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-04 14:21:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"55926d3022f108ba6a81139c18aefe7ed5d7c6a8","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-02-04 14:21:10.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e BTW, push the patch with topic \"bug/1913180\"\n\nDone.","accounts_in_message":[],"_revision_number":3},{"id":"cdddc76c9283170e871a9f21af03eb0bdae41b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:08:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afbed495e1d44e8b7a4141ebeb7c5a2 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6bc8d776d6940f9953bed2ae38e5bdd : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c04acd3e0c8a463da0ad13199c358f29 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52640bcf172d444f94bae1299ea70c44 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a45ea04aa0402f9a97fa83a1466de8 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/832f3e28ddd2435490a9fce210125d43 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/223a052049e74a358855e87aff863444 : SUCCESS in 4m 08s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b049fe9cfc844c8ea9a31b6a5ebbc346 : SUCCESS in 1h 48m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a92f31836c034a9da7fca3848e43ef35 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f7dc063254f39e5e0f771e8cada5b3b48168e4d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-05 08:16:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5336d5041d83fabd444d28973fd79d404ee4b9c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 08:16:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6324da4e2b9e48740785d9277a2c91564dadbf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:26:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57db34322b9d45cda8946a109c3a1a95 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7993330458f446968bcb5460ab2c9297 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da15e3ad0fd943b199b07c49bd4a25b9 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8b403cf3f74f50be8928a9c8e911a5 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab3da93109b4e56a3b9960e5155da39 : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce998bc3a2784c1aaaabf09bfddcf55a : SUCCESS in 1h 07m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f389c313bfd48e480e8191c0ce4ac97 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":3},{"id":"25a62345da4180c01f9ddd0e0e8a3c846911e0b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:26:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e646e8136a27450bbbcdbe01deccd14c6f0e04a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:27:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5227ac909f8348108e3cd53baca1dfa2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2dabcc5cf7d68bf2a4640f07d5e170aa8b911390","revisions":{"ce74632492f485bdf3c51eaa2d8e91c69cf9957f":{"kind":"REWORK","_number":1,"created":"2021-02-04 13:27:35.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/83/774083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/774083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/1"}}},"commit":{"parents":[{"commit":"4632e29a41ce08c485d5724971dce9ac55c6a7d7","subject":"Introduce rbac-address-group api extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4632e29a41ce08c485d5724971dce9ac55c6a7d7"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 12:27:27.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 13:26:23.000000000","tz":180},"subject":"Add qos rule validation for network","message":"Add qos rule validation for network\n\nMinimum bandwidth QoS rule is only applicable for the network\nwhich is backed by physical networks.\nWhen updating network policy on network without port\nwill raise exception.\n\nPartial-Bug: #1913180\nChange-Id: I6b57f7ce6d411be388c63b87f2c75f2de703cc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ce74632492f485bdf3c51eaa2d8e91c69cf9957f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ce74632492f485bdf3c51eaa2d8e91c69cf9957f"}]},"branch":"refs/heads/master"},"c0c4f5cc65243052313a6404e88da202d6d4a81f":{"kind":"REWORK","_number":2,"created":"2021-02-04 13:29:55.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/83/774083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/774083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/2"}}},"commit":{"parents":[{"commit":"4632e29a41ce08c485d5724971dce9ac55c6a7d7","subject":"Introduce rbac-address-group api extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4632e29a41ce08c485d5724971dce9ac55c6a7d7"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 12:27:27.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 13:29:30.000000000","tz":180},"subject":"Add qos rule validation for network","message":"Add qos rule validation for network\n\nMinimum bandwidth QoS rule is only applicable for the network\nwhich is backed by physical networks.\nWhen updating network policy on network without port\nwill raise exception.\n\nPartial-Bug: #1913180\nChange-Id: I6b57f7ce6d411be388c63b87f2c75f2de703cc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c0c4f5cc65243052313a6404e88da202d6d4a81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c0c4f5cc65243052313a6404e88da202d6d4a81f"}]},"branch":"refs/heads/master"},"2dabcc5cf7d68bf2a4640f07d5e170aa8b911390":{"kind":"REWORK","_number":3,"created":"2021-02-04 14:16:39.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/83/774083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/774083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/774083/3"}}},"commit":{"parents":[{"commit":"4632e29a41ce08c485d5724971dce9ac55c6a7d7","subject":"Introduce rbac-address-group api extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4632e29a41ce08c485d5724971dce9ac55c6a7d7"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 12:27:27.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-02-04 14:15:02.000000000","tz":180},"subject":"Add qos rule validation for network","message":"Add qos rule validation for network\n\nMinimum bandwidth QoS rule is only applicable for the network\nwhich is backed by physical networks.\nWhen updating network policy on network without port\nwill raise exception.\n\nPartial-Bug: #1913180\nChange-Id: I6b57f7ce6d411be388c63b87f2c75f2de703cc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2dabcc5cf7d68bf2a4640f07d5e170aa8b911390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2dabcc5cf7d68bf2a4640f07d5e170aa8b911390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
