)]}'
{"id":"openstack%2Fneutron-lib~774929","triplet_id":"openstack%2Fneutron-lib~master~I3b82ffe4fa41f1b57ffbad88a86455930bdf1b32","project":"openstack/neutron-lib","branch":"master","topic":"bp/address-groups-in-sg-rules","hashtags":[],"change_id":"I3b82ffe4fa41f1b57ffbad88a86455930bdf1b32","subject":"Add shared attribute to address groups for RBAC","status":"MERGED","created":"2021-02-10 18:45:31.000000000","updated":"2021-02-15 19:09:00.000000000","submitted":"2021-02-15 19:06:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774929-bp/address-groups-in-sg-rules","meta_rev_id":"b009de0d1dfc9c7f1289a75eef19249e061eddf7","_number":774929,"virtual_id_number":774929,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 19:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-12 11:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-12 08:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-02-12 11:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 20:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 10:09:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-02-12 08:29:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-12 11:06:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a1ad91587927dc6fd19b0aaa1511bf7f9618227f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 18:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed1dc8e642441fd8c3b2ce3c67b3be36a30e42c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 18:46:25.000000000","message":"Topic add-shared-attribute-to-address-groups removed","accounts_in_message":[],"_revision_number":1},{"id":"00f2f250dc58d4d7bfa51fdf3036e94b066955b1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 18:46:31.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":1},{"id":"564eb75366103b9b2f1c201e76a8755244ba262d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 20:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6c5a70c54c41fa906ec436c974b1b2 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea5bf13b388a4402ba5918f3ebf31498 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8e1ae348737436aa0406839dfc04a27 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e1b61af9e74ede91b8599d1aae4516 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e37cbe6b47f453a8a417877cdd9de6a : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696122c2ae644952a3066e54cae986ae : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8af43bad85b446e1929c3d96591fa486 : SUCCESS in 4m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2f2dd148c9d406f849f2aae5dd3f484 : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b9f2d51b7004771b82bc41b5771f466 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4096a02843bf5659a5180f10e4ab0291078f9cf1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-11 10:09:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3e1dd9976a1b279f53924f97a049bf9b49e8bd9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-11 20:56:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e53b2cc6a967e847297d74fd109afbb7d4373df3","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-12 00:35:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"195f0503e2d339f5345df3a6419a5763cc618420","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-12 08:29:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b96907cc8c673e225791abdcd36d4f9456c16642","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-12 11:06:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f70ea61b2bd2bb1e54add1a070a73e78145662b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 11:07:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b36e4266cc6ff5f40b2086fe3c55315cf42a32f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 12:39:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285cccd0bfe14d509277951f158a73ee : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e00e00a68024826b6330464eb8ff99f : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21dc1aec0d564d95a0b11832b373fffc : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d3a22e68f3440eebf7a024019d84119 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ceb6c3ed3604419962b7d82eb1f48e3 : SUCCESS in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4f67296885f4931a6578bf449bb4e11 : SUCCESS in 1h 27m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dd76ab51b1ee42b8b10610612575f6e7 : POST_FAILURE in 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6c34eee5863e139b17510ea8a9709ac5b803dca8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-12 14:03:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab13b2d8c54918ddb79b1e6e8420f29f1f00ad0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:39:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a564f4eeb4461dac4578599d903103 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e357a4e6a314e2da14cae961303b820 : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58686b6f5144406aa711b09f5af4fc18 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a3abe11c9a437b9ef2881c7237e8b4 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa4dd7f92beb42a4a54b3ca91d621597 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c653acc3e35643caa3df1d2b94d821b3 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65dcdede92a4442c807716047eaf0066 : SUCCESS in 4m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e22d8b0ffc14d7fb79ccf5f6c9c90ec : SUCCESS in 1h 33m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eb9a852d42e5484eb8e041d29c672908 : FAILURE in 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"523f57479ba68799efc73322dcea6dff37a4b229","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-15 11:11:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c0710253f471d16ccbe03a93a44dc5952387a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 13:20:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a134ca59713847d4b53dca3eacc96dda : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fc56a27faa94b04850fdcf46f6d852a : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebf1314af5bb4ffba75faa86974e31b6 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086a53dbf5ce4d1f89db87d837ae71e8 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec6d8032fea4854932a4ed41ab17a3f : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01ab88a7e62e45f9b5c379ee2c094cd4 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd801641f7ab4bc080c2266d955dbac1 : SUCCESS in 3m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1d1174bac944481843d1a58eb7fe8da : FAILURE in 2h 05m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b4a5cc5ff99f42ae9feeda0322c1830f : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dc91812a18625bf788669c99c01120d1abb30168","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-15 15:15:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27d19aa1e9fc866585b5ccb291feaf7933e88392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:09:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b059f83017410488fc28a75cdc99bc : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126629d64b41482cb6e63cbea61c42b5 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b67b6e25134943b8a6b88780c56ed73b : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e219dafd76bd45c69d49ca4ff8f880a6 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/208d5b2cba434a3299aa83f756eae4a0 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e14c66117de43e3bdbd33d02ea06b5e : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fd26ba0959f4c4780438a4624025658 : SUCCESS in 5m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c4c1906963940338fc5e185c0da6301 : SUCCESS in 1h 41m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a631bcdb46a14b0eba8df88bb70b0b4a : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0ed013722848f84ecf4847fa4bc8171425c7d0af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:09:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4593fc4295eea380c5d29605d747c14b126ed980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 19:06:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e0c2c83ba443748fb195680167ddaf : SUCCESS in 15m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2556d7c9a8174557ac43f06ebdf5a6e7 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca37e56f949c4c3888eaa01cb6be5860 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2469a55d30a84c578e896bb7a6766113 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f22c10d714240948455b6c01e3fce17 : SUCCESS in 6m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad9c0b3d83634796b60ca902c33f3dea : SUCCESS in 1h 50m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64b7582eea384e74abe6b4b582b81725 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"515dd238ee4f87be48e370dbba782a673440e582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 19:06:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b009de0d1dfc9c7f1289a75eef19249e061eddf7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 19:09:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab294e51fa8a4911b554ba9d0e38b2b9 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4","revisions":{"e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4":{"kind":"REWORK","_number":1,"created":"2021-02-10 18:45:31.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/29/774929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/774929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/774929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/774929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/774929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/774929/1"}}},"commit":{"parents":[{"commit":"455f6244762c593e311795741e7a3b3a164b3147","subject":"Fix address groups API definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/455f6244762c593e311795741e7a3b3a164b3147"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-10 18:45:09.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-10 18:45:09.000000000","tz":-360},"subject":"Add shared attribute to address groups for RBAC","message":"Add shared attribute to address groups for RBAC\n\nIn order to be able to share address groups with \u0027access_as_shared\u0027 RBAC\npolicies, the \u0027shared\u0027 attribute is added to their API definition.\n\nChange-Id: I3b82ffe4fa41f1b57ffbad88a86455930bdf1b32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
