)]}'
{"id":"openstack%2Fneutron-lib~776656","triplet_id":"openstack%2Fneutron-lib~master~I75d04306f334475e329b124cb47c513421c2438a","project":"openstack/neutron-lib","branch":"master","topic":"minimum-bandwidth-allocation-placement-api","hashtags":[],"change_id":"I75d04306f334475e329b124cb47c513421c2438a","subject":"Add PLACEMENT_REPORT plugin constant","status":"MERGED","created":"2021-02-19 15:22:09.000000000","updated":"2021-02-22 10:24:21.000000000","submitted":"2021-02-22 10:22:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776656-minimum-bandwidth-allocation-placement-api","meta_rev_id":"13aa68602643aa20a0c618d9a48cb4f1e118780a","_number":776656,"virtual_id_number":776656,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 10:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-22 09:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-20 07:04:41.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-02-22 09:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 17:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-20 07:04:41.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-02-22 09:02:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ddfcf375b44a31e32a6f5ffe7ae79eb96cb3059b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 15:22:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7eb777d85af44f288f22689f56c74cb304b1385f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:18:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8df5b816ec45359455a00071b8216c : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a3aaef471aa4227b13b74387abab6af : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37c43a47c4ec40ef9881e3fb7b2cb97b : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4049fb5116704c5cb0010faa08d6ac9e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6301f9a90504d8e9cd00bfdd5fc130d : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6e0b7197b214ec5904d8839d5f1ed3b : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c417f8b4c77d40f184b1a9a9303b3729 : SUCCESS in 5m 16s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be7803c5d8f4487e8b32e6c48041749d : SUCCESS in 1h 49m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0f409d68f3c4fd4b4ce61eecd5b37b9 : TIMED_OUT in 1h 00m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7114ace27ae84e069b8f83733c4c0e30 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"67be925ae6f200c11dbac04aecf28aaee143d75e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 17:32:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c81c1c30f3238361a95bac1d37ac70b69ca2d558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:16:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36fadd0e9daa4144b3e75ad66f40a47a : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d3e45a7534c44038c1f9dd434c92e72 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4feccbecb9c345f09275901df7ecc69d : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c670fbdb68d4f7d839a81bae601a93c : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e568140b89463ba468a31ec518b89e : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce590c5d5f24334a15bfc215dcd0d6f : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/798b277c8bb3427483cb26b0566d423c : SUCCESS in 5m 00s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62b1fe5aa9094872ac09a253e5883294 : SUCCESS in 1h 28m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/295a3f9d2f924431bcd9094444110ca7 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb681ebe059e411caf25893467248f96 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7fb4cb95ad055fad1b4b218cee04ccb1c6b62469","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-20 07:04:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"867b3683844d4d192b2a74751cf951dce5736d29","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-22 09:02:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0a43cc96c5adf0b8eec9b8689f69de8178c4688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:02:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57757f23112c381cedc6db18be43db96ff12b189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:22:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb248956e3aa4888b2c7d323d428b994 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/372281d5051b460ab88a86bfd00b46d7 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192beaf773d64682b38e1d43c95ee6be : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58b43efeb7234b6db40ecbe6011d5509 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3891385506bf4081a1eeed61d3167354 : SUCCESS in 5m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aecb89ea27a945a39e29eb57ce7178f4 : SUCCESS in 1h 18m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1eacf92a10f40368fa0700ca365482b : SUCCESS in 4m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0f5a1b119ad348e492dab172a1dbe000 : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0f36dd6c94a0566ac5fd901dba89924039c0d428","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:22:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"13aa68602643aa20a0c618d9a48cb4f1e118780a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:24:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559f271fabdb445da1f6b49f9e50ea85 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262d57fef05e4fdfbca36286b1d72e86 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"698e4c8daa7d43018a71122ec5b0cd5b17b55141","revisions":{"14637c56e0c94f830628cc1ed5fe45dc58ef875d":{"kind":"REWORK","_number":1,"created":"2021-02-19 15:22:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/776656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/776656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/1"}}},"commit":{"parents":[{"commit":"e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4","subject":"Add shared attribute to address groups for RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 15:15:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 15:17:08.000000000","tz":0},"subject":"Add PLACEMENT_REPORT plugin constant","message":"Add PLACEMENT_REPORT plugin constant\n\nDefine a PLACEMENT_REPORT constant so that it can be used elsewhere\nrelated to the ``placement_report`` service plugin.\n\nPartial-Bug: #1578989\n\nChange-Id: I75d04306f334475e329b124cb47c513421c2438a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14637c56e0c94f830628cc1ed5fe45dc58ef875d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14637c56e0c94f830628cc1ed5fe45dc58ef875d"}]},"branch":"refs/heads/master"},"698e4c8daa7d43018a71122ec5b0cd5b17b55141":{"kind":"REWORK","_number":2,"created":"2021-02-19 17:32:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/776656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/776656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/776656/2"}}},"commit":{"parents":[{"commit":"e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4","subject":"Add shared attribute to address groups for RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4f10dfa3e0c9da8fe5297e2c24957b6cf559fb4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 15:15:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-02-19 17:32:45.000000000","tz":0},"subject":"Add PLACEMENT_REPORT plugin constant","message":"Add PLACEMENT_REPORT plugin constant\n\nDefine a PLACEMENT_REPORT constant so that it can be used elsewhere\nrelated to the ``placement_report`` service plugin.\n\nPartial-Bug: #1578989\n\nChange-Id: I75d04306f334475e329b124cb47c513421c2438a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/698e4c8daa7d43018a71122ec5b0cd5b17b55141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/698e4c8daa7d43018a71122ec5b0cd5b17b55141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
