)]}'
{"id":"openstack%2Fneutron-lib~779515","triplet_id":"openstack%2Fneutron-lib~master~I53726d64167ba38c233c2cfc095ff6462bcb664e","project":"openstack/neutron-lib","branch":"master","topic":"bug/1918274","hashtags":[],"change_id":"I53726d64167ba38c233c2cfc095ff6462bcb664e","subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","status":"MERGED","created":"2021-03-09 15:47:02.000000000","updated":"2021-03-10 11:25:49.000000000","submitted":"2021-03-10 11:23:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779515-bug/1918274","meta_rev_id":"e50667e5dfe916319e330e39918ff6609ab3dd64","_number":779515,"virtual_id_number":779515,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 11:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-03-10 09:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-10 09:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2021-03-10 09:02:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-03-10 09:38:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 15:47:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2021-03-09 17:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 07:17:22.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-10 08:58:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-10 09:38:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"47bb73fd7a04cbab811a979ea65c58d00a04193f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 15:47:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcf3071dbf46279d5b0596939bda6d00c2171e38","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 15:50:55.000000000","message":"Patch Set 1:\n\nHi @Miguel. I have a question related to this patch and [1]. What is the purpose of [1]? Because if we merge this patch and we define a physnet only with the name (that means all VLAN ranges), this check will fail now (well, not fail but \"vlan_ranges\" now will be [(1, 4096)], instead of []).\n\nIs that correct?\n\n[1]https://github.com/openstack/neutron/blob/a9fc746249cd34cb7cc594c0b4d74d8ddf65bd46/neutron/plugins/ml2/extensions/dns_integration.py#L366-L367","accounts_in_message":[],"_revision_number":1},{"id":"6bf8ad1c72bedf969fb52695c639ce3e2f8f9694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 15:51:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5619726832467e947c5c1cfb8ab4f3c52905d136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:25:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430d8721d160428aab5112cc1c4d8b9a : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f051e3fa390649308f3a02c9299c885d : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/666b16573c8d4028b76ab7024285a868 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11eb40a4c1e4ab58223f452693e407a : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/354234b027684bb3b41fa95e91db19e3 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaeebb6b044c4f23a3a10bd21ddc46d9 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237bda2b0a6447758d338d31289b9a7a : SUCCESS in 6m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6364e014bbfa4302a3ebf75744274c9d : SUCCESS in 1h 28m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/100631d10c0e412aa73a08c9ef68d1fc : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d528200867a9cd9a0631abec78465a9927429488","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-10 07:17:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dc800cfa095da3b7e10128a43ce63e086f1f12d7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 08:58:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9fdb58b121ad1cc75a9149ef787d60ccfaeac4a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-10 09:02:31.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"1d18e16e48e19db0b57db6f6be7a1f1555358b4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 09:14:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc6c2d0d7417ef903d0cb112ac0a5b1d35d28c6c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 09:38:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6de509849e0006525c3a0bfa9df929092711ed86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 09:38:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6377babbc2c92aceeac629efbef13c02a9ff9f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:23:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2313340824241b9b69e04b543b8686b : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73c194dfaac2408da559d5feb810a500 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e36e7b11f54479b9c640cc688cc565 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81883c9c683c492fbd72cbcadb5fd7ba : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8430f70d29984c1f9d3af8128575aa87 : SUCCESS in 4m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18f510e91c19401ca48d733980940f3f : SUCCESS in 1h 37m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/531c642ebf5f424dac58835f41a7981a : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a69ce894e835cb9e5b82dd104de82751fffee356","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:23:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e50667e5dfe916319e330e39918ff6609ab3dd64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:25:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe13b8fb16948fa8864cccc7e1c5b02 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02e070fe099651ad5abea87819c7d3e729885130","revisions":{"c819afc7f8d017f673300c471f25f9fbc0a9bbb7":{"kind":"REWORK","_number":1,"created":"2021-03-09 15:47:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/15/779515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/15/779515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/1"}}},"commit":{"parents":[{"commit":"698e4c8daa7d43018a71122ec5b0cd5b17b55141","subject":"Add PLACEMENT_REPORT plugin constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/698e4c8daa7d43018a71122ec5b0cd5b17b55141"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 15:41:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 15:41:17.000000000","tz":0},"subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","message":"Populate full VLAN range in \"parse_network_vlan_ranges\"\n\nThe method \"parse_network_vlan_ranges\" parses the\n\"network_vlan_ranges\" parameter that defines, for the VLAN type\ndriver, the available physical networks and the assigned VLAN\nID ranges.\n\nIf only the network name is passed without defining a VLAN ID range,\nthat means all valid VLAN IDs are allowed (1, 4094).\n\nThe returned variable should contain this range in this case.\n\nChange-Id: I53726d64167ba38c233c2cfc095ff6462bcb664e\nCloses-Bug: #1918274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c819afc7f8d017f673300c471f25f9fbc0a9bbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c819afc7f8d017f673300c471f25f9fbc0a9bbb7"}]},"branch":"refs/heads/master"},"02e070fe099651ad5abea87819c7d3e729885130":{"kind":"REWORK","_number":2,"created":"2021-03-09 15:51:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/15/779515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/15/779515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/15/779515/2"}}},"commit":{"parents":[{"commit":"698e4c8daa7d43018a71122ec5b0cd5b17b55141","subject":"Add PLACEMENT_REPORT plugin constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/698e4c8daa7d43018a71122ec5b0cd5b17b55141"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 15:41:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 15:51:05.000000000","tz":0},"subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","message":"Populate full VLAN range in \"parse_network_vlan_ranges\"\n\nThe method \"parse_network_vlan_ranges\" parses the\n\"network_vlan_ranges\" parameter that defines, for the VLAN type\ndriver, the available physical networks and the assigned VLAN\nID ranges.\n\nIf only the network name is passed without defining a VLAN ID range,\nthat means all valid VLAN IDs are allowed (1, 4094).\n\nThe returned variable should contain this range in this case.\n\nChange-Id: I53726d64167ba38c233c2cfc095ff6462bcb664e\nCloses-Bug: #1918274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02e070fe099651ad5abea87819c7d3e729885130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02e070fe099651ad5abea87819c7d3e729885130"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
