)]}'
{"id":"openstack%2Fneutron-lib~779720","triplet_id":"openstack%2Fneutron-lib~master~Ib2933e4f52e8d7585c13a0d1a3485389fa37b432","project":"openstack/neutron-lib","branch":"master","topic":"bp/enginefacade-switch","hashtags":[],"change_id":"Ib2933e4f52e8d7585c13a0d1a3485389fa37b432","subject":"Remove warning about session with and without enginefacade","status":"MERGED","created":"2021-03-10 11:59:22.000000000","updated":"2021-05-27 11:37:10.000000000","submitted":"2021-05-27 11:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779720-bp/enginefacade-switch","meta_rev_id":"ed6c236d835981887f586fd64a7f72806cb42ca3","_number":779720,"virtual_id_number":779720,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 11:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-26 13:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-25 10:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-05-24 19:58:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 08:46:02.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-05-26 13:26:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 12:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 12:39:20.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-03-10 15:25:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-25 08:27:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-25 08:27:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-25 10:52:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"4e536164f8641aa9b5975c3b5abed1c607ef2a44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 11:59:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dec10ed2817a3ee819bb9210dcd09a2794bae889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:32:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ebafd8a63f49918d650c5297276081 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041bb1610a0c42b1844254e308aa278d : FAILURE in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/521bafe6744949d3aed699ff176f371a : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cde58d52254c8bb368605fa3613ab4 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b7f3527e944cfb8417ec458e7f18c7 : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6bae49a87b45459433a5ec0ba0b2f6 : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64a1c566621647dd9e51a151310e1353 : FAILURE in 5m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a536de5d84d45f4afae5a0c01e3ddbc : FAILURE in 27m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8adf89771bef4a36adfb0a19a2d636d0 : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9a4c25c8bd3542496aa5b8a6cb28e0d033e041ac","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 12:39:20.000000000","message":"Patch Set 1:\n\nI thought we had an open review existing for that but I don\u0027t see it anymore (and it definitely did not get merged, as code is still there)","accounts_in_message":[],"_revision_number":1},{"id":"e122b26d07bdf3b0b2bf640c9d2974e37a031c1f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 12:54:45.000000000","message":"Patch Set 1:\n\n\u003e I thought we had an open review existing for that but I don\u0027t see it anymore (and it definitely did not get merged, as code is still there)\nFound it https://review.opendev.org/c/openstack/neutron-lib/+/761729 not sure why it was abandoned though (because of the resulting test failures?)","accounts_in_message":[],"_revision_number":1},{"id":"8fcf8aae39433487c4ccf8afedbfccc7ce767923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 15:20:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b63b7a773d88954fc54b9ca4c680ad69f1979e62","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-10 15:21:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e I thought we had an open review existing for that but I don\u0027t see it anymore (and it definitely did not get merged, as code is still there)\n\u003e Found it https://review.opendev.org/c/openstack/neutron-lib/+/761729 not sure why it was abandoned though (because of the resulting test failures?)\n\nUps, I didn\u0027t saw it before. But as this one is now opened, if You don\u0027t mind we can go with it. Is that ok?","accounts_in_message":[],"_revision_number":1},{"id":"05059bda51b499235206856775eae373bbaf6df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:13:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de36c027e844ae29b75ea8cd0d67af7 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e84445e38de04d249728dd33fa8dd1b1 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06afa6fc0847454e8bb17f4f4171b91e : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a319a08ae2b44c19d55e5484305555f : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f77fba2b175945a9a20aa37987ec98f9 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a835580d03264765b67c2be51ad57326 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1101e482bbaa48db871ca116ff14b0f7 : SUCCESS in 4m 52s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5d0f7d971f14b37aa29b2aa3add8102 : FAILURE in 41m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/77e254f97ca0430faa125227ac710ba1 : FAILURE in 31m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a925edfb65323822ff9a6c8236f160d9ba6ff670","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-12 08:49:44.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e Ups, I didn\u0027t saw it before. But as this one is now opened, if You don\u0027t mind we can go with it. Is that ok?\nNo problem at all, I was just wondering why previous attempt was abandoned, maybe because of these in tempest run?\n\nCallback neutron.services.l3_router.l3_router_plugin.L3_HA_NAT_db_mixin._precommit_router_create-1579995 failed with \"set_extra_attr_value cannot be called out of a transaction.\",Callback neutron.services.l3_router.l3_router_plugin.DVRResourceOperationHandler._set_distributed_flag-1585543 failed with \"set_extra_attr_value cannot be called out of a transaction.\",Callback neutron.services.l3_router.l3_router_plugin.RouterAvailabilityZoneMixin._process_az_request-2096603 failed with \"set_extra_attr_value cannot be called out of a transaction","accounts_in_message":[],"_revision_number":2},{"id":"986a502011db600cce3766839a6553ae138e8bdc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-12 10:02:48.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e Ups, I didn\u0027t saw it before. But as this one is now opened, if You don\u0027t mind we can go with it. Is that ok?\n\u003e No problem at all, I was just wondering why previous attempt was abandoned, maybe because of these in tempest run?\n\u003e \n\u003e Callback neutron.services.l3_router.l3_router_plugin.L3_HA_NAT_db_mixin._precommit_router_create-1579995 failed with \"set_extra_attr_value cannot be called out of a transaction.\",Callback neutron.services.l3_router.l3_router_plugin.DVRResourceOperationHandler._set_distributed_flag-1585543 failed with \"set_extra_attr_value cannot be called out of a transaction.\",Callback neutron.services.l3_router.l3_router_plugin.RouterAvailabilityZoneMixin._process_az_request-2096603 failed with \"set_extra_attr_value cannot be called out of a transaction\n\nThose errors should be hopefully addressed by https://review.opendev.org/c/openstack/neutron/+/779721","accounts_in_message":[],"_revision_number":2},{"id":"fd1ded252ceb642c1fa228b44c490bb6284a41a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-08 13:52:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"092d261043e5f644e8d1c097c4145c2b2d0a6604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 14:36:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f614157ceac5462dbefad2de9564a845 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d3ad08bf0f4f6ebe4cf8f80623b711 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b34bc7a2d44131954a0e42141f631f : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2a579be4c4420aaccbed9a05b522e8 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bff3f46ffeeb4918a97aabd3a2420bd5 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c8f1bf9645d4d7d9f9757daba2628fb : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2575fa1d0a4340b4b5c025bd35685f : SUCCESS in 5m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18231d1c080b468586e010f2e8cff74b : FAILURE in 31m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d02d73a10f074c6e9d133db8adea703b : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":3},{"id":"5a3ff4dce0b416b6bb7370ef1949deb1d08da2a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-08 17:06:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4a06c3e771f115462f7cc1530bc9a8d8ae209ae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-08 17:07:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9a9866c3e4a84cede4c58be80a2720b8b049e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-16 12:45:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a77fc8954dca177209aa15c6882408a106de373","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-16 12:49:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b71f15b7ae5113b919bab8750cb92a94ec50c141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:16:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65076b34c24484ca37b5d2c2b1fe3c3 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebc6794cc34f485cb11279cb0e73d6dd : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d1dc3d8b36841f79167844b94092936 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93bdfbd14600436493364011be81b2e6 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62a6ef9f35fd455ca6c831eb1f99b415 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f28d79f9f1043619c72179721c45412 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60955f9676354b68ab7a3e688f824534 : SUCCESS in 4m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4526ee29598749c1aec850f695ae8dcb : FAILURE in 25m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59b0864edbc546beb4e3927097ce7741 : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":4},{"id":"34867e5aa052669d81330a9a1c6fcfa49cbd44ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 14:29:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4277505067c4a4420df708148e71db0a4cc3df17","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 16:50:18.000000000","message":"Patch Set 4: Code-Review-1\n\nHmmmm, this is not working: https://review.opendev.org/c/openstack/neutron/+/786642","accounts_in_message":[],"_revision_number":4},{"id":"7e1e301aac45cf5fafc6b71dbadbb25e61293667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 13:58:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d8665b8022528ea1aec765703b9ffae6b8239432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 14:00:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f4bb8f1bed5aaf3e657438f973f84490273c0706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 15:36:14.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052b0b7bd402404c835c42b1558ba16c : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ea4bc5a4aac42959eee7df19e11febe : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e246b03b89594b3ebbc4e6a7b4bd627d : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a58f9a125941ab98f5d3e5398def68 : FAILURE in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c410b2d5b774efea147aeaa64c69973 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0cf914f64194be9a5777ceac74ae2f8 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5b7aefe5d0245a3947de9c4eadaa537 : SUCCESS in 5m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d5aa0308d99496496d2f191617e7389 : SUCCESS in 1h 31m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/66d2f29eaf664c28bf4daf54a1075f70 : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":6},{"id":"afa118ee45e54c5ccb443242e482cfa9c948896a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-21 13:16:33.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"0cb564fd34a76fdfab88ce6fc708b13d7aca2a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:14:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d881dd0a849405eb1a8f90b64cb1381 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dec4b8091f3441ca94df6405022936c6 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6f84a2ef7854216bf083d8c5e9ff4be : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a5bb2a88f3461f94e98853756f9f60 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60709913f1d344938c5524a956a61678 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f3862215e92439db4e0d307d2f91541 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6680d1d3e70412bac7697d04a8eea90 : SUCCESS in 5m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a76b4b035564cc19d73b6506002311c : FAILURE in 1h 15m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/63318116d2814ba8b1b2b9bc3d2a6bec : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":7},{"id":"185cbb40a4dda61e944422f73a14d17cc3ff1aaa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 08:40:01.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5845de9ab98af953803858dea28bf79fba61ebbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:08:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4345a0f093437282fbcb14eb0caa65 : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9083e824e173470599c730d9f6151898 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea3a293504074fdbad83e31d28fd89fb : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01d97428e66436d9d3ce680b159b8f2 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f3fee9353b4b508099108f78780ac2 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae51d228eb84650aa811d82e5a08635 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38da5191c8b24fa9b986191d1234aaef : SUCCESS in 4m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/971bc4cd11c54ee4999965724b9c1257 : SUCCESS in 1h 26m 48s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f85554b911a3422b9bb7b9a2dce5eef9 : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":7},{"id":"7d3b7846461197f0a0c3b7e38d363baf6df0d2ac","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-24 19:58:38.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c88f88147a6031adf595d18b3e692a5147b4ad5e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 08:46:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"20265f11206b43dadd02bde5a351ba496e99b6d4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-25 10:52:03.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"db29f26243d0ef653cadf05b2bb2448b063f09ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-26 13:26:35.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"387ad1ff72ccff77e49b8eef0a24a81f7f2cbdf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:27:03.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a8c16ceab89a3775f214c3e9ee7248b1fbb2d6f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 15:54:59.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac6b808a6924acc9bc1645dbaf5e01e : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69d0cf0b8fac48e3a3b7113e0b2286db : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132cf6d0d4c14d408b2b0ec1557e657a : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11aeeff8873444e9a165cbff2dfc022b : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77202f5b4250457bbaa001ab53802ff8 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7781302385be4620a4c5eabf0b1f7023 : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa814558305a4ec797cc31a74c44815c : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":7},{"id":"de5dbe313b209985e13f955cad3278beced8abf4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-27 08:35:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"054d6222d0581aa045a919aef43f3af5001058b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:01:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a498cadbb249418861e56af765daba : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66f68125002844f29b5d35ac00bfb9fa : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00de5dc3861405dadc318af67c7a9d2 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580b8f26890a48ef9638f68dadc7143d : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14cf54a183b24d6aa44ef60cc3b51df0 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3db8ad092d7c4255ba43e36a2eb63e21 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb8d37f79bc242abbb5f5f7e4ba11aec : FAILURE in 22m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc1897d6051a4d9180d027d939afce9b : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6f47917583b4d60816c24c69e49bb86 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":7},{"id":"2a49023802c368d4c7896802f0060b07774b35ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:01:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1e3ed7998b3ec7a0a23dacbcc6590f1d4e85020f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:35:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d52b56db424ea3a333c4ccd6c83813 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/223c22eab5b942ccb169f28e4288b7ac : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aad528cc8e2498fada7de9818796531 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc3d4e36099476f8b0377b9b48cc32e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd5a9242a9964a3da858ea647e8b1e0f : SUCCESS in 6m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2d64d586c68462ebd9124652b91dc45 : SUCCESS in 1h 33m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cd35b82289f04d6a832350c396850a64 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":7},{"id":"3e6651946c96a563a8c89865aa8abb63753f0c3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:35:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"ed6c236d835981887f586fd64a7f72806cb42ca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:37:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575306eca54044f6ab170b96ea7be98e : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a05778013cf25d2f70bd97662ea93b1efca49dd9","revisions":{"05e1a111308738a7894befd25b5f8bb1365b1758":{"kind":"REWORK","_number":1,"created":"2021-03-10 11:59:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/20/779720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/1"}}},"commit":{"parents":[{"commit":"02e070fe099651ad5abea87819c7d3e729885130","subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02e070fe099651ad5abea87819c7d3e729885130"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:59:18.000000000","tz":60},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/05e1a111308738a7894befd25b5f8bb1365b1758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/05e1a111308738a7894befd25b5f8bb1365b1758"}]},"branch":"refs/heads/master"},"48860e5b0e56cf04f38dc834858ada95d1773a5f":{"kind":"REWORK","_number":2,"created":"2021-03-10 15:20:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/20/779720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/2"}}},"commit":{"parents":[{"commit":"02e070fe099651ad5abea87819c7d3e729885130","subject":"Populate full VLAN range in \"parse_network_vlan_ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02e070fe099651ad5abea87819c7d3e729885130"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 15:19:33.000000000","tz":60},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48860e5b0e56cf04f38dc834858ada95d1773a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48860e5b0e56cf04f38dc834858ada95d1773a5f"}]},"branch":"refs/heads/master"},"cd474ada939daa1f257f2bc3c038376fb24761cf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-08 13:52:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/20/779720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/3"}}},"commit":{"parents":[{"commit":"f01b2e9025d33aeff3bf22ea2568bda036878819","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f01b2e9025d33aeff3bf22ea2568bda036878819"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-08 13:52:06.000000000","tz":0},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cd474ada939daa1f257f2bc3c038376fb24761cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cd474ada939daa1f257f2bc3c038376fb24761cf"}]},"branch":"refs/heads/master","description":"Rebase"},"b47ac08ce50f25cf22159ad26adad49f759a3704":{"kind":"REWORK","_number":4,"created":"2021-04-16 12:45:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/20/779720/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/4"}}},"commit":{"parents":[{"commit":"f01b2e9025d33aeff3bf22ea2568bda036878819","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f01b2e9025d33aeff3bf22ea2568bda036878819"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-16 12:45:21.000000000","tz":120},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b47ac08ce50f25cf22159ad26adad49f759a3704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b47ac08ce50f25cf22159ad26adad49f759a3704"}]},"branch":"refs/heads/master"},"4bdeab04e97ce5abd9e5df4814666da35e47ae4e":{"kind":"REWORK","_number":5,"created":"2021-05-18 13:58:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/779720/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/5"}}},"commit":{"parents":[{"commit":"8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a","subject":"Merge \"Move bgp api-def to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 13:58:28.000000000","tz":0},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bdeab04e97ce5abd9e5df4814666da35e47ae4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4bdeab04e97ce5abd9e5df4814666da35e47ae4e"}]},"branch":"refs/heads/master"},"c60d94a0c19a1f05b2e54f834317d13556564a2b":{"kind":"REWORK","_number":6,"created":"2021-05-18 14:00:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/779720/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/6"}}},"commit":{"parents":[{"commit":"8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a","subject":"Merge \"Move bgp api-def to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 14:00:13.000000000","tz":0},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c60d94a0c19a1f05b2e54f834317d13556564a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c60d94a0c19a1f05b2e54f834317d13556564a2b"}]},"branch":"refs/heads/master"},"a05778013cf25d2f70bd97662ea93b1efca49dd9":{"kind":"REWORK","_number":7,"created":"2021-05-21 13:16:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/779720/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/779720/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/779720/7"}}},"commit":{"parents":[{"commit":"8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a","subject":"Merge \"Move bgp api-def to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-21 13:15:48.000000000","tz":0},"subject":"Remove warning about session with and without enginefacade","message":"Remove warning about session with and without enginefacade\n\nNow we moved all code to the new enginefacade so we don\u0027t need this\ncheck anymore.\nAnd BTW. I think it wasn\u0027t working properly as even if we are using new\nfacade we still see plenty of those warnings in e.g. logs of the\nneutron functional tests CI job.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib2933e4f52e8d7585c13a0d1a3485389fa37b432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a05778013cf25d2f70bd97662ea93b1efca49dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a05778013cf25d2f70bd97662ea93b1efca49dd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
