)]}'
{"id":"openstack%2Fneutron-lib~782318","triplet_id":"openstack%2Fneutron-lib~master~Ief57a797a1b1ed34f1d7ec9d7d9f983451647696","project":"openstack/neutron-lib","branch":"master","topic":"upstream/feature/fip-pfw-port-range","attention_set":{},"removed_from_attention_set":{"30695":{"account":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"last_update":"2022-03-07 21:19:10.000000000","reason":"Pedro Henrique Pereira Martins replied on the change"}},"hashtags":[],"change_id":"Ief57a797a1b1ed34f1d7ec9d7d9f983451647696","subject":"Add port-ranges on portforwarding api","status":"MERGED","created":"2021-03-22 18:57:16.000000000","updated":"2022-03-07 21:19:10.000000000","submitted":"2021-04-09 09:03:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"782318-upstream/feature/fip-pfw-port-range","meta_rev_id":"e316f9f4ae2d85a0bff21b60c2ea550b4e409920","_number":782318,"virtual_id_number":782318,"owner":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 09:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-04-08 10:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-04-08 11:53:34.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-04-09 07:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 20:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 08:14:16.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-24 13:16:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-07 01:02:01.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"CC"}],"messages":[{"id":"3210aef4dc5cb6331ca763ac829fe6f14c236d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-22 18:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a3fcfb8f2f609af2588245cf959dcd896e353bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:19:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4a30f8495dc64d4e84a45faab52ed9ed : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9541220117e5476a9cc0df87210a8e99 : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec993d48b3fe4d7fb9fda5e9db2b67ae : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/038a1f70941e49528c3f3d9a0545065a : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07237b938d56454ea234774ae823213c : FAILURE in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb4dd33feed409286118af5f2b30e0d : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c05cbabb23cd4215bd70a435dcd27067 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a152bfd41a549b2ae8c707ca8b7c4f3 : SUCCESS in 8m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fc50a0de1ee4f7bb3495ecb866356fa : SUCCESS in 1h 20m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aafbbd190dd4a75919cf6ab092ed639 : SUCCESS in 4m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e25dbeb92ef8480b8f092f4a65dc061b : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b5493feb6470bb846cac0d306f9952fad0c2175b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-22 21:15:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe329eb491a3ba790d3589a218e8a84014d66b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 22:46:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9384583e128a4c908ab03bffcee2d6c8 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8513571a1ea34d7fb09148b256942292 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29446727464a4978aa3037faa475e1f6 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/612c9c60fe634630ba876bda6dda7c7c : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ee894814da400bb977ea5b7d98f732 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5e9c7d814eb4a4e976f979274e4e061 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb83d0cf1de8472fad40c7954ece464a : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6ba60ab1264f4db728542d8a25e942 : SUCCESS in 7m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba59ec035476429f851efb600414c017 : SUCCESS in 1h 21m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc0de5c33744f4cb68ee824d60ab3ca : SUCCESS in 6m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bdf9e8a457fb433596b99b38ca1d88c4 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6ae40d7fdb85d507be6d4484b34f7c67e21041cb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-23 08:14:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c49a0b89b7e6741e1d5640d66b1238a927ab82a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-23 14:17:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58f90334032841611baf116367c55a47f246d002","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-23 14:18:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"472c3f82b8df7a711c84596dc8139bd6313610bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:24:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c36553dd1444fc2abe6db94fbccb185 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df9fb50e0814cf797cfeb4226c1b897 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b24ec90785a46ec9278cfc7a7d3a4c7 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64cd172b2adb4c5cac4e0b9ed1e8b6fb : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c6876470434461b9c59c1445dca268 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0568806736a54d3f87f3199dd706f58d : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7f95c3ae6941c58e9aa366f0b83296 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a4e985322ee48988bbc3f6b4f01b049 : SUCCESS in 4m 54s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad4e35e287694708aeb7fcbdf3059fa7 : SUCCESS in 1h 03m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1208ab2323a4dd5893322e93b913829 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db83e6731d734f06b4f5adafb51a8f29 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2033ce617eeceb177cacac66fec3962eab948829","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-24 13:16:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5d80aee61e10eadc3db7781d90466cb674844f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-24 14:56:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7bf5a26ce02eaa267b870b7d2a860cde03f913e8","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-24 15:07:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad0f4c67ea94866364aac364b970f3edd1d765ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:08:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bf901cb5b6a40509224dbaeebca73ff : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bc5560195b4e53a93bb8ab1c7d233a : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bf90d93f9a94deb8b4107e225c0ff92 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3912c9f86d8946a2a99f5cf1d7f31411 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf587776bb7438889cf40c4c67fbe25 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2702bda908a94d13bf40cc5086fa819e : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/800b44a5a8f74ef183a416131239cf08 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d73e9aaa2ec40fb889a80a1c5536c51 : SUCCESS in 6m 01s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eda8fbed1c4a4f98a27358ebd1581af3 : SUCCESS in 1h 55m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0432f409fc12414e9dfdb248aa3288ed : SUCCESS in 6m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b61ee85b933c473fa0e1b8b879ad09d4 : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":4},{"id":"2ae83fa0df54236740d94d3a6107ce1c6babde43","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-25 08:32:17.000000000","message":"Patch Set 4: Code-Review-1\n\nSorry that I didn\u0027t check it earlier but You should also update API-REF: https://raw.githubusercontent.com/openstack/neutron-lib/master/api-ref/source/v2/fip-port-forwarding.inc","accounts_in_message":[],"_revision_number":4},{"id":"bd23bc883b2ca57abdfa7797029985b0fd09a1bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-25 21:20:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"785323450e0666bd64f78ecea13ca812b439699b","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-25 21:23:16.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Sorry that I didn\u0027t check it earlier but You should also update API-REF: https://raw.githubusercontent.com/openstack/neutron-lib/master/api-ref/source/v2/fip-port-forwarding.inc\n\nNo problem Slawek, thanks to remind me about the api-ref 😊","accounts_in_message":[],"_revision_number":5},{"id":"24cfb7db8403b057da26c7125074e0924119f9ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:56:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cba40057252a458d96ae2bc9d0d19e81 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82d09d153cf45f3b99c2cb221bfbe05 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f24ebf431cc14451801f18f00529c37a : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db4afc31853745288947ce66769a18de : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c87075c0ff145d0ac0cbda8020336ab : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e203f7b3424469ab6b93dd5bd2482e : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d392b639a2404e50ab8b884e85c432d8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74106c431781487f91855d1d73e8ae44 : SUCCESS in 4m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d45eda6227bd4775b73f4740c8f225e3 : SUCCESS in 1h 34m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cadd6093ed84245b57ef653c9487349 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cbe6b1b5aacc4cb49fe78cd39576a23c : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":5},{"id":"88701e960e0eccf8fda6034da46d708b57569b5b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-26 08:13:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4f28126060416b7011ff1342e32a34f2a402b14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-30 12:30:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09d18a6174c8fe05e26c4d6b3cdf3f52a790b2f5","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-03-30 12:31:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e2c7c5b95803dfcfbf712748aa0bd679fa3added","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:08:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/14dfe2eea2d54c5b9c507140a338030f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6256dd30254b47c2b7d1d3a8631176d0 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dd5b93fe7848ea8bd4166524cabfc8 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f8def39409f4bf2a3e6440c1a93f0e6 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82fa42ee8e0425b9ebb4fb738abc8aa : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b14821dd441466f987f3069b54442c7 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/279ee489208642789ae66ec5850eba50 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cad5aa18cae4f8490be9bbd034faad8 : SUCCESS in 4m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/064215bdc6e14c94a7a9286077f5e534 : SUCCESS in 1h 32m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7012ec9c8527484d9c9eac40972b3f7b : SUCCESS in 4m 18s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ff6b3b39b9b4dfab6388ef591878e2d : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":6},{"id":"daf9cdfc1e9e8dc47c58cfa52c72f2f5390db0da","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-31 07:29:54.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"206a9684973f7346ca4d971806990a8b9a51a276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-04-06 12:49:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c1492c5e795494a45cffb247d6a15a155828e2f5","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-04-06 12:50:37.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"49c8f2bcf6028eaec50daa00543ce4eb4ffad082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:42:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cc721e0224f24223899d3a569f3cf1dc : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e779f7c02c4c1e9d2b5f2ad405348c : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0da91662b5634ddc94340f55c11cbe89 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f98fd774bbd4e2a86ad7b0f99a63dca : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1d43245e6c46f3abe3929d36df517d : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c32d52587e14e02a430b573600953f2 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cb7e34c196a4c349ad2cb639f12c3a5 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621a26f838ad4bb59ca90850d7d02ce7 : SUCCESS in 5m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b38751a91e614649b841c6507afde3ac : SUCCESS in 1h 50m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/524dbd767b314349891a6b5e871ed7fd : SUCCESS in 4m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/510bd810ee2d4802996cda049e68b557 : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":7},{"id":"2b4ba9aad5e3e95ba281f9d6d278d84ca4b1715d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-08 10:12:12.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a0185158c95271bc53a572194fd13e00dbc93163","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-08 10:52:48.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bf4e6cb37f29afb13fe5c7204f3b0e42e374ddfd","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2021-04-08 11:37:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab2dc81ae53b09a8901fd13bb720a60b0f0590fe","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-08 11:53:34.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3bf3307a0838ae323a2388f3eaf5fc03af669fa9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 07:36:09.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e33dec93df9dfb01af36f4bd7afacf9a38ee9ae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:36:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"93f224ba4f2914b3c4dc9416c92305da946a902c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:03:27.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1f07e8c3582f4f989ecc9e23f85c12f8 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c144634ac054ec5ac9bd61413261137 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a33476e3ef5c4d34a3f0cd052093c678 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969a9312ebed410eb099033b1c054e07 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/265767af6e434b0eaa5ebb43a4b366f7 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e143bcfaba054f17829709aaaf6c1fe3 : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d82e36965f7d47dc9ca911a042e73bec : SUCCESS in 1h 22m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44eaa0fc51064570b32d108941d4c2e6 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c08afa254557442da693326fbaddb42f : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":7},{"id":"6022b42b62f38211d0a5bbcb6ceb807fdf002876","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:03:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"4f6341f9d2ff5043f2a821c50a30461ae9f196b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:05:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2804a141198646d095cd344d2b8243e9 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6c6e5b80594e5ebca41d1d57ae5ea4 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e2cd2b9e474a09ba2c730e9eca7023 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7},{"id":"243828dc3747911b363a2c6dc048ff1c245d8b38","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-07 01:02:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e316f9f4ae2d85a0bff21b60c2ea550b4e409920","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2022-03-07 21:19:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"69f4c6399d952afa30612f604ee8edf7b4778ad4","revisions":{"79f4f56ac0cb4481372600de26cdd55765586081":{"kind":"REWORK","_number":1,"created":"2021-03-22 18:57:16.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/1"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 18:43:37.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nCloses-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/79f4f56ac0cb4481372600de26cdd55765586081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/79f4f56ac0cb4481372600de26cdd55765586081"}]},"branch":"refs/heads/master"},"c22b989a6d3f9923534036d856c385378c2495dd":{"kind":"REWORK","_number":2,"created":"2021-03-22 21:15:08.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/2"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 21:14:54.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nCloses-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c22b989a6d3f9923534036d856c385378c2495dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c22b989a6d3f9923534036d856c385378c2495dd"}]},"branch":"refs/heads/master"},"fc539af5b48e0739ab24c41414dda50631494e8d":{"kind":"REWORK","_number":3,"created":"2021-03-23 14:17:38.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/3"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-23 14:16:31.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nCloses-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc539af5b48e0739ab24c41414dda50631494e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc539af5b48e0739ab24c41414dda50631494e8d"}]},"branch":"refs/heads/master"},"cfb8dcf61c0b39cc5e6f804ab7e7099d2befac99":{"kind":"REWORK","_number":4,"created":"2021-03-24 14:56:32.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/4"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-24 14:55:09.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cfb8dcf61c0b39cc5e6f804ab7e7099d2befac99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cfb8dcf61c0b39cc5e6f804ab7e7099d2befac99"}]},"branch":"refs/heads/master"},"9e85d0178d2646c914d13f71c38b9f2fd6786f44":{"kind":"REWORK","_number":5,"created":"2021-03-25 21:20:13.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/5"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-25 21:19:37.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e85d0178d2646c914d13f71c38b9f2fd6786f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e85d0178d2646c914d13f71c38b9f2fd6786f44"}]},"branch":"refs/heads/master"},"6cb46ca99aee4a15e86840b6e9c8bec457c49489":{"kind":"REWORK","_number":6,"created":"2021-03-30 12:30:13.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/6"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-30 12:30:00.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cb46ca99aee4a15e86840b6e9c8bec457c49489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6cb46ca99aee4a15e86840b6e9c8bec457c49489"}]},"branch":"refs/heads/master"},"69f4c6399d952afa30612f604ee8edf7b4778ad4":{"kind":"REWORK","_number":7,"created":"2021-04-06 12:49:56.000000000","uploader":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"ref":"refs/changes/18/782318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/18/782318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/18/782318/7"}}},"commit":{"parents":[{"commit":"1361d8cfa4350d26e1d12393000a276176052bff","subject":"Merge \"Fix policy enforcement for network mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1361d8cfa4350d26e1d12393000a276176052bff"}]}],"author":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-03-22 17:55:08.000000000","tz":-180},"committer":{"name":"pedro","email":"phpm13@gmail.com","date":"2021-04-06 12:49:40.000000000","tz":-180},"subject":"Add port-ranges on portforwarding api","message":"Add port-ranges on portforwarding api\n\nThis patch is the first of a series of patches\nto implement floating ip port forwarding with\nport ranges.\n\nThe specification is defined in:\nhttps://github.com/openstack/neutron-specs/blob/master/specs/wallaby/port-forwarding-port-ranges.rst\n\nChange-Id: Ief57a797a1b1ed34f1d7ec9d7d9f983451647696\nImplements: blueprint floatingips-portforwarding-ranges\nRelated-Bug: #1885921\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69f4c6399d952afa30612f604ee8edf7b4778ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/69f4c6399d952afa30612f604ee8edf7b4778ad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
