)]}'
{"id":"openstack%2Fneutron-lib~785590","triplet_id":"openstack%2Fneutron-lib~master~I46f0cc0bbf680fbceff1806720bb40b60a5e54e8","project":"openstack/neutron-lib","branch":"master","topic":"rehome-ovs-constants","hashtags":[],"change_id":"I46f0cc0bbf680fbceff1806720bb40b60a5e54e8","subject":"Rehome ovs related constants to the neutron-lib","status":"MERGED","created":"2021-04-09 10:25:53.000000000","updated":"2021-04-15 08:42:43.000000000","submitted":"2021-04-15 08:41:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"785590-rehome-ovs-constants","meta_rev_id":"3dbcbb014d5c3b27ac035082858b3f4abb4bed76","_number":785590,"virtual_id_number":785590,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 08:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-04-13 06:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-04-13 09:13:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-04-15 07:05:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 11:45:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-09 12:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 01:17:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-13 09:13:32.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"81c19dade2e720ee59a7d0b9613e9b9ee2dfc456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 10:25:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32bb20e573a96ccf86f59571f8aab97943c05810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-09 11:42:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb28bcf3813b0cd765421510717d5d267f7831f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-09 11:45:58.000000000","message":"Patch Set 2: Code-Review+2\n\nGood idea","accounts_in_message":[],"_revision_number":2},{"id":"5bd6e199f1291fe1d638dba9b1b6b6cc2adfdb71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-09 11:46:38.000000000","message":"Patch Set 2:\n\nMaybe you can add a release note.","accounts_in_message":[],"_revision_number":2},{"id":"76a38fd6c0387ccdec7891a0fb5eb539abdcfa90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 12:56:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360126a9eb85484788085d82337bc958 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13fdcaab61b44744ab4c6b456c6b262b : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5ee27cc21f14240a06349adb305ebc1 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742e8671c0de43d6b199155f03c7acfa : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5af73519d75d438092adb53f9d629eee : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7383ede7e2d84843b7684aa17a43f864 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd0d3ac3ada34f1297afe8033c5b7673 : SUCCESS in 5m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d0cd771122e45d0a5c33203409c7c62 : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/250f8fa8563a48c48db6818bf1b42da4 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7b80d15d4faf12da68b4d92fccbdb61bdcd4319a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-13 01:17:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fb02764720944e65175f80d519cfe3a854f80f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-13 01:18:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56db81ca7078203454ba8bcc24e7567959d204b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-13 06:24:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4a3ac35d7551aac78ba44082fc1a11d8fa362eb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-13 06:25:16.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Maybe you can add a release note.\n\nGood idea. Done :)","accounts_in_message":[],"_revision_number":3},{"id":"14a30351d21f1bd989300a9c8125fc6765d1152d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-13 06:26:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b5458628b16b6b4d6f64806b2c19f060a968606c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-13 06:48:32.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the release note","accounts_in_message":[],"_revision_number":3},{"id":"02c45a647d413004ac66e3a04e01a083348bbb1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 07:46:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ee3ce85736484285ca196acd7fafa0 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d76494bd57d4db9ba1aca85ccf46c52 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959bbe86132643f79fe77e2b11af8133 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78950572d93144ec89f5db7b6af4ea25 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/645a7c31976d40adafed8252c8e18317 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba96a490ece42018c9e6b8be894135a : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0ed7fb246f4b5596c8af59316b3720 : SUCCESS in 5m 27s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/925aefaee2564df892b589ed6f0831f5 : SUCCESS in 1h 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01c961ebe0cc49079af1e8ff497c86a6 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61a4e81af6e24c2b9ad56e601a2e5434 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8035b41fa99ecc4af6323ec3742f2d1419fbeed2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-13 09:13:32.000000000","message":"Patch Set 3: Code-Review+2\n\nOne step before me","accounts_in_message":[],"_revision_number":3},{"id":"5c63dcf7632ba785ea2b8c5a6a29f4133256e5f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-15 07:05:48.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76eb635fdaa0e4923ec34bf6b8ab2ab4179726a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 07:06:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"020d80d8a2283992245bc1d118a3ddc602ae3fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 08:41:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df03a2602e44400aae674cbb64eefc77 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b77bcacacb7d4007b632844727bddde0 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b40426d37e74784a72e0ec07e5e6b73 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d48817d5e70f4a139b4897a86354f466 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/063341f4fb354f37b09a494dde54c5cf : SUCCESS in 4m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc2a1872e365486e832af98bd68e1f82 : SUCCESS in 1h 29m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d62cf22c12e14e8188be992e2e8eaa13 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62f826929a0745ec9fc3734b83ef9e70 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":3},{"id":"08c059d2590c88b8f6d2253b6fc574979f43b1a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 08:41:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3dbcbb014d5c3b27ac035082858b3f4abb4bed76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 08:42:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64b45ebf9d64e42874505e9ad4b61b2 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cc4484a99e54c67b493d37ec39180f8 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"14dd8af92d0d536b5e034d49c69201631b28f4fa","revisions":{"ca154997d7a91e087f500420cebff5fdd1128eea":{"kind":"REWORK","_number":1,"created":"2021-04-09 10:25:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/785590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/785590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/1"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:22:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:22:53.000000000","tz":120},"subject":"Rehome ovs related constants to the neutron-lib","message":"Rehome ovs related constants to the neutron-lib\n\nConstants were in\nneutron/plugins/ml2/drivers/openvswitch/agent/common/constants.py\n\nbut at least some of them, like TUNNEL_NETWORK_TYPES,\nPEER_INTEGRATION_PREFIX or FLOOD_TO_TUN are used in other projects also\nso it would be good to move have them in the neutron-lib.\n\nChange-Id: I46f0cc0bbf680fbceff1806720bb40b60a5e54e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ca154997d7a91e087f500420cebff5fdd1128eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ca154997d7a91e087f500420cebff5fdd1128eea"}]},"branch":"refs/heads/master"},"90cceb18794e2735bf1b5e802b7ff65857c1a01f":{"kind":"REWORK","_number":2,"created":"2021-04-09 11:42:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/785590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/785590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/2"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:22:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 11:42:14.000000000","tz":0},"subject":"Rehome ovs related constants to the neutron-lib","message":"Rehome ovs related constants to the neutron-lib\n\nConstants were in\nneutron/plugins/ml2/drivers/openvswitch/agent/common/constants.py\n\nbut at least some of them, like TUNNEL_NETWORK_TYPES,\nPEER_INTEGRATION_PREFIX or FLOOD_TO_TUN are used in other projects also\nso it would be good to move have them in the neutron-lib.\n\nChange-Id: I46f0cc0bbf680fbceff1806720bb40b60a5e54e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90cceb18794e2735bf1b5e802b7ff65857c1a01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/90cceb18794e2735bf1b5e802b7ff65857c1a01f"}]},"branch":"refs/heads/master"},"14dd8af92d0d536b5e034d49c69201631b28f4fa":{"kind":"REWORK","_number":3,"created":"2021-04-13 06:24:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/785590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/785590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/785590/3"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-09 10:22:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-13 06:23:32.000000000","tz":120},"subject":"Rehome ovs related constants to the neutron-lib","message":"Rehome ovs related constants to the neutron-lib\n\nConstants were in\nneutron/plugins/ml2/drivers/openvswitch/agent/common/constants.py\n\nbut at least some of them, like TUNNEL_NETWORK_TYPES,\nPEER_INTEGRATION_PREFIX or FLOOD_TO_TUN are used in other projects also\nso it would be good to move have them in the neutron-lib.\n\nChange-Id: I46f0cc0bbf680fbceff1806720bb40b60a5e54e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14dd8af92d0d536b5e034d49c69201631b28f4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14dd8af92d0d536b5e034d49c69201631b28f4fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
