)]}'
{"id":"openstack%2Fneutron-lib~786174","triplet_id":"openstack%2Fneutron-lib~master~I030d4c1db0dfb8f0a242cb60d641d7ed05ee8051","project":"openstack/neutron-lib","branch":"master","topic":"bug/1923700","hashtags":[],"change_id":"I030d4c1db0dfb8f0a242cb60d641d7ed05ee8051","subject":"Add ExternalDNSOverQuota exception","status":"MERGED","created":"2021-04-14 08:08:40.000000000","updated":"2021-04-14 17:27:06.000000000","submitted":"2021-04-14 17:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"786174-bug/1923700","meta_rev_id":"60b881dbd155d6c096678e4a0a1cda93eb91aa65","_number":786174,"virtual_id_number":786174,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 17:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-04-14 15:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2021-04-14 12:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-04-14 15:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 08:09:01.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-04-14 08:34:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-14 11:02:05.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-04-14 11:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 11:28:33.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-04-14 12:24:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6ecb25fe7ca75e56d3650a5716bb7ff4b5b61b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-04-14 08:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a36bca7503800e2b9a24b21f754752c0deceac55","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-14 08:34:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e89a49cbc05e4dd46a25a3fc5516791791321e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-04-14 09:30:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"67ab620b59f623157e6d3d9ff5bbe58dfc299c38","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-14 09:50:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f52ad9e58a8632b3655bd5f4004cbdbcde10d9b0","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-04-14 11:02:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6f07c9424c339ae21ef5273e2a0a563db135682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1066351b83af4da4bfabfc2274ad503b : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4935433f58bf4b09b4612eaaba580dd4 : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/427769388cab45c1a949058a5d74025d : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b821af62a6451fa1a5e2019a31a805 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39e3182f3506498a9d69b2e6df3e7b9e : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2a066ad885c4c1282705d79deb5404d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ba8ca1f37bb4a6983b72b400325a824 : SUCCESS in 4m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a282151e65f34b0dae565f427b064527 : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0565d33f7e21423f9785f8beed707f71 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a2f95098abdc634cc6a150a98bd2ab9c31c2e151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-04-14 11:41:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d08724ce38bcd3830cda414dd0266cc87fe06848","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-04-14 11:42:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4182bd383c8e08c90f16a9cd1d023537a077a7a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 12:24:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"145abfc3318b101aab5ab987b83d7f3693f67f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 13:09:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c54b3ec1d7a4140be2f959a16fb4ef2 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ea26af1a3c94a37917eff9d49c44041 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343489877b564bd491b2437980acd7b9 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb992f0e1410498ebb2bdb217e7cee68 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38597b2087b406ca74b15e15fcc0f76 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/494e239d0cba4df3bacbc93594fc1b88 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2abfa17e25a4ce9a37741b41dcc916f : SUCCESS in 4m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a88269a54074606983dbbcc28696903 : SUCCESS in 1h 22m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0edb9a8daec4f13b5d9c3797c59ad0a : FAILURE in 39m 44s","accounts_in_message":[],"_revision_number":3},{"id":"f2a0943869bdd66ca0f62d033dc0965c11df6a31","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-04-14 13:20:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2251003f3de8619970bef611bda7466517e5ea40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:23:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4792d33f66814c7a9a42a449b1d68f31 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0b2e96dc9440f7817923d40a4a901f : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0c2c4528fdf4e059479463fbccbcdae : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca21cc066010454295eb996f68eee062 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6852a21b4888434394a73c3b75b5890a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f36f4afcaf479d8e4d45a11a2c0011 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/578d30c8091c4604bd898503829c8ecf : SUCCESS in 5m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e25caa66330a4552b681d163ab06986e : SUCCESS in 1h 19m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01b02c37790f400f83f4b358b10183c1 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":3},{"id":"5027429384ec8f34fd309ed34c9fbab018db6c3c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-14 15:53:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f89f98d288ba53717008bec72953330b7d08b1de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:54:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"72733219be531801eb6e318392995fdfe90e6f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:24:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5518ca64d7c84a738f4bae2ea55c39c8 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddcbd34081234befa404ce41136364fc : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8e70536d9c4a7fb403e5f691fb8564 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa9971710a3493699eb105dab0b7e1a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc31298b62cd4be3b1eecc99faed3bc5 : SUCCESS in 4m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/647037a7a0a646c193dea080c83c74f6 : SUCCESS in 1h 24m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/06550a11bc9948618ee1828b37b1a10b : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":3},{"id":"464152c5bc4378e0a1ade5f558a7787073542fea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:24:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"60b881dbd155d6c096678e4a0a1cda93eb91aa65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:27:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb0a40389914124aaa5f39f6bc465ce : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7566924e0e43cc7532177c8a0cb95b9f93bfcc50","revisions":{"edaa34a45c5f9d0535c28b4c70b7f65db1d87a37":{"kind":"REWORK","_number":1,"created":"2021-04-14 08:08:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/74/786174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/74/786174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/1"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 05:30:00.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 08:06:23.000000000","tz":120},"subject":"Add ExternalDNSOverQuota exception","message":"Add ExternalDNSOverQuota exception\n\nWhen floating IP is created and the overquota limit of the external\ndns service (e.g. designate) exceeds, recordset creation fails. Report\ncorrect error/exception in neutron instead of HttpException.\n\nCloses-Bug: #1923700\nChange-Id: I030d4c1db0dfb8f0a242cb60d641d7ed05ee8051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/edaa34a45c5f9d0535c28b4c70b7f65db1d87a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/edaa34a45c5f9d0535c28b4c70b7f65db1d87a37"}]},"branch":"refs/heads/master"},"cb62931c730209040a26238849046db420c0fa93":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-14 09:30:22.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/74/786174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/74/786174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/2"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 05:30:00.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 09:25:27.000000000","tz":120},"subject":"Add ExternalDNSOverQuota exception","message":"Add ExternalDNSOverQuota exception\n\nWhen floating IP is created and the overquota limit of the external\ndns service (e.g. designate) exceeds, recordset creation fails. Report\ncorrect error/exception in neutron instead of HttpException.\n\nRelated-Bug: #1923700\nChange-Id: I030d4c1db0dfb8f0a242cb60d641d7ed05ee8051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb62931c730209040a26238849046db420c0fa93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cb62931c730209040a26238849046db420c0fa93"}]},"branch":"refs/heads/master"},"7566924e0e43cc7532177c8a0cb95b9f93bfcc50":{"kind":"REWORK","_number":3,"created":"2021-04-14 11:41:59.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/74/786174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/74/786174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/74/786174/3"}}},"commit":{"parents":[{"commit":"968b979c2a8864e894e0e6d40fdf329806fd21cf","subject":"Merge \"Add port-ranges on portforwarding api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/968b979c2a8864e894e0e6d40fdf329806fd21cf"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 05:30:00.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-04-14 11:36:46.000000000","tz":120},"subject":"Add ExternalDNSOverQuota exception","message":"Add ExternalDNSOverQuota exception\n\nWhen floating IP is created and the overquota limit of the external\ndns service (e.g. designate) exceeds, recordset creation fails. Report\ncorrect error/exception in neutron instead of HttpException.\n\nRelated-Bug: #1923700\nChange-Id: I030d4c1db0dfb8f0a242cb60d641d7ed05ee8051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7566924e0e43cc7532177c8a0cb95b9f93bfcc50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7566924e0e43cc7532177c8a0cb95b9f93bfcc50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
