)]}'
{"id":"openstack%2Fneutron-lib~788729","triplet_id":"openstack%2Fneutron-lib~master~Ibc3b9ffd01fe9eeb6a78671bd286ae3ade8816fa","project":"openstack/neutron-lib","branch":"master","topic":"lazy_loads","hashtags":[],"change_id":"Ibc3b9ffd01fe9eeb6a78671bd286ae3ade8816fa","subject":"Allow lazy load in model_query","status":"MERGED","created":"2021-04-29 12:21:51.000000000","updated":"2021-05-27 14:47:42.000000000","submitted":"2021-05-27 14:45:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"788729-lazy_loads","meta_rev_id":"dbdfa3291d715c4845aa0b477d6965ee7abd4047","_number":788729,"virtual_id_number":788729,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 14:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-25 17:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-04-30 07:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-07 12:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-26 13:25:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 13:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 16:32:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-30 07:08:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-07 12:15:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a5c6973ac44d9f9b4018aa07632399c8b8ca8027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-29 12:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da5d5e4a25794e5c30af699b7340b613eb4ae981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:35:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f759a361f08f4cb8bb7cd74173c2d28b : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8508f67f90048c688f96b1f0d191176 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba49919e7c804380871978a2a7982d2d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8157292b2eca4c8b80dde0d665661dac : FAILURE in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db17a7bdd4a4a36b8113906e57584dc : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef4b2f04ad504c198575565d9c0bb1ab : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83c17ab4be4947caa9780c02cb99aeec : SUCCESS in 5m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc5e2f553bd747eabdc7a782c739eb09 : SUCCESS in 1h 12m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e12d7e6b92844cd6bdb1e5048a3f4a4c : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8ef481563a6c34db2c9928044a1f0bc99635b317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-04-29 15:02:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"366e8f24e8516349f29076c5bf41f3871d17118d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 16:40:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cef01292cd4ae1af47fa468fd9eb3b : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76b618f7de6d46059d8b6ec18917e09f : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82f2733c90d643e1b1f054f739e4c845 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c666e49a284ab09ab2706d62e9fbd3 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66669fb7d8a64e52808e43f6818b2a66 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/277a9c50fae74afc8b17bd3f8de0e0c4 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48a2d8f01ef048ed9a2672e87abb83dd : SUCCESS in 4m 12s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e52832152a74891b16a896044ceee6e : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dace4210204741c2b2b0fa8d406c564e : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":2},{"id":"96973b85f0aee411dafd2cc7f5f13bf1528876c9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-30 07:08:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d8ef9f1ac80a7211865a874961a7af02131726dd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-07 12:15:02.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like a good idea","accounts_in_message":[],"_revision_number":2},{"id":"e470ffe93601c8a0768f600fa4dcbfbef04cf0a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 14:06:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1f0d28d65ababacf709ae093efb8eae82cd26a9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-11 09:53:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e07997c838a844d0c49b63a2d0b1644b062efab9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 07:03:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73871ef8b1abe1dfc4e8273b92203d3d637550d1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-13 04:39:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff68707b515ae6d6acebbe16059d37440fd8c742","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-21 15:02:56.000000000","message":"Patch Set 2:\n\n@Rodolfo, please check my last comment, does it make sense?","accounts_in_message":[],"_revision_number":2},{"id":"07c1c8696692d4008c15458307cc92b267749763","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-25 17:10:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2c8801be14a5b80a5ca2b5541e98d9ca944861cc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-26 08:56:10.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nSeems safe to approve :)","accounts_in_message":[],"_revision_number":2},{"id":"4b6b94ac4c680fb161cb5ff98582ea95093420fc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-26 13:25:59.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"039c91e3ca45121c3db4434c200d286f00d6d92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:26:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"025425e6f7e0e248ee6a66c26063fc15012694bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 15:08:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bce65f71c5049cab339c01747172891 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd19f0f3f97e4c1c847e79cc56c3a440 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29c26728b9c413189a05a1efa1b0884 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82799d961c004cc2877337d4ba0e4ab0 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24178e977131451990424d0773db89ab : SUCCESS in 4m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c656dd4b81443bdb2eba253b8700584 : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4ebb32be39e043039bdf0db64b613961 : FAILURE in 47m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1ca1deaec4133b751fd5edca6a1fa4264fb493c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-27 08:35:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"86764c8e8125e5bb8e9e8e2a6461abba3fe1c2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:16:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3df507d00544ef9eefde36925a9cfa : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e08e61efe241158dcd6e53024aeb93 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc3e46359d75454ba248723e4a037169 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa57e9e57077470c81598bcd9f71cd10 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cffb6be0eabf4fa3b77e903412d8d378 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1001ee6e03cb489db320d461abdc0767 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b4b26bcd155470fab0e3c02d32b0a38 : SUCCESS in 4m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37d9358681824dfca0117eef9ae71a8e : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73464dc4bcf74fd0a99a9dd8bd81a480 : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":2},{"id":"06399027b9ea49002d9d515fd421b02766bba21f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:16:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c424ece4c3a7a1139b9b51bd4e223095c20ff75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:35:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cca72725d6d4076a515fe97eb0d1f6d : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdfee9d5d25e4ab18921ae0b967feca6 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3946c02c84bf422988f5fda7c2feb2c7 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5db0480f0b5645f294d7a9f77bc387d8 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/014eddabb37d4059b8cd135a83b40bd1 : SUCCESS in 4m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d168f1e8c6b48c281f8420f5e4e3d07 : FAILURE in 1h 07m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61eced2c8aca4075a4118f9f4867f762 : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"eaf493531ff8aec9a0b4a7dcf50b1c734e429240","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-27 11:55:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"79911c730cb1fc3d6d6692bfa74f3d1cffc55f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:13:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f02b7326704f1ba73a0f2c2a15f6b9 : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1e9a4a934404c20b2ac3ef528d6d2ac : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69c05b94dab54f1e8b04b5ba4b5bea3c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315c1c17478f4bdaa509a6d13101cd88 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b5eec5b3ae46de825f3df96fc5be81 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71712799745f4c3aa655518fd75618ac : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73630098ee1f4c2290c9fb58f6583232 : SUCCESS in 4m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef20bcc865414c2c821cc24db51534ee : SUCCESS in 1h 16m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68d9281df9bf43daafe2d1cbb145ecf8 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":2},{"id":"89a9164e7a2c461bfb66993e83bd4b771cedda0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:13:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b399a18cb8be9698547e1c1ac9b07b0848969ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:45:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42baebc15308427ab6776f9ea96febd6 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27429d3aafc8418cb651c93ec181b1c4 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4855cf770dc142baa63a2234a5858fb2 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1f5b150ee994bdb9ec750887e9e1d3d : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e6c9bc70068460b94fbd5ba4cf9d8e8 : SUCCESS in 5m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0275765f0b9e4c6dae9d7d29eae847aa : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/79d8109efdab4188b23c4942c5fd8486 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a411aa07d6cfe058f2f3bfef14bdb8729f9187e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:45:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dbdfa3291d715c4845aa0b477d6965ee7abd4047","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:47:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca6427133ee424aba62aa468c87a943 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a165d9608047ec97a270050bf2ce2978d368152c","revisions":{"8f198fb3fbacd6fc3770ff2422f2575777b367c6":{"kind":"REWORK","_number":1,"created":"2021-04-29 12:21:51.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/29/788729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/788729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/1"}}},"commit":{"parents":[{"commit":"ba36d60717b85d715e08556ffca23446431402eb","subject":"Fix pylint\u003e\u003d2.8.0 error R1732 \"consider-using-with\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba36d60717b85d715e08556ffca23446431402eb"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-04-29 12:17:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-04-29 12:20:23.000000000","tz":180},"subject":"Allow lazy load in model_query","message":"Allow lazy load in model_query\n\nLazy load may be useful for performance if some fields\nare not always required in all cases.\nThis will be used in network and port list operations\nto override default joined and subquery loading.\n\nChange-Id: Ibc3b9ffd01fe9eeb6a78671bd286ae3ade8816fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8f198fb3fbacd6fc3770ff2422f2575777b367c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8f198fb3fbacd6fc3770ff2422f2575777b367c6"}]},"branch":"refs/heads/master"},"a165d9608047ec97a270050bf2ce2978d368152c":{"kind":"REWORK","_number":2,"created":"2021-04-29 15:02:54.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/29/788729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/788729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/788729/2"}}},"commit":{"parents":[{"commit":"ba36d60717b85d715e08556ffca23446431402eb","subject":"Fix pylint\u003e\u003d2.8.0 error R1732 \"consider-using-with\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba36d60717b85d715e08556ffca23446431402eb"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-04-29 12:17:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-04-29 15:02:40.000000000","tz":180},"subject":"Allow lazy load in model_query","message":"Allow lazy load in model_query\n\nLazy load may be useful for performance if some fields\nare not always required in all cases.\nThis will be used in network and port list operations\nto override default joined and subquery loading.\n\nChange-Id: Ibc3b9ffd01fe9eeb6a78671bd286ae3ade8816fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a165d9608047ec97a270050bf2ce2978d368152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a165d9608047ec97a270050bf2ce2978d368152c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
