)]}'
{"id":"openstack%2Fneutron-lib~790285","triplet_id":"openstack%2Fneutron-lib~master~I33f42ec487567327c7c96afd81dd61e9e091c08f","project":"openstack/neutron-lib","branch":"master","topic":"bug/1927763","hashtags":[],"change_id":"I33f42ec487567327c7c96afd81dd61e9e091c08f","subject":"Add \"cache_ok\u003dFalse\" flag to TypeDecorator TruncatedDateTime()","status":"MERGED","created":"2021-05-07 17:27:31.000000000","updated":"2021-05-18 14:00:30.000000000","submitted":"2021-05-18 13:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790285-bug/1927763","meta_rev_id":"9b6d226b8a8b9e1e31d9d71ebdba150a1e469b01","_number":790285,"virtual_id_number":790285,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 13:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-12 19:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-11 08:20:26.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-05-12 19:47:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 19:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 05:37:48.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-12 19:47:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7c598616f1fb75581990e40c2fa250765cbe1aaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-07 17:27:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b14618c91a34c7f8955d1179d78108453d90232e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 19:07:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d13e68a5efa4badbefd75cadefffd5b : SUCCESS in 7m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57e0abc29315463cadc6f81d69d02bf5 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab2b448d53524565b1cc6fc4775e5873 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4337baf6e962405bacd3357f4fb1f7c6 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b15f40b2ce046e3b0ae9ac9fc6c3a78 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75cffecf854141f1bb9de9954357d0a4 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102ce943da574deaaf3a53c32c06e55d : SUCCESS in 4m 49s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30b38044f8a2457b82d9647a69710499 : SUCCESS in 1h 34m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62400ced72f740d5bd291bc11255a9d5 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":1},{"id":"222788fd94ed859c56e5bd7809479ce5138b3e29","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-11 05:37:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7c914959351a65dbbacd6623dbce9e239def556","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-11 08:18:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9061ec4c1eac25617dbab252766a0fdb1936a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-11 08:19:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eac773a21969aa4e3b2cc4423407fce0445dadda","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-11 08:20:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d51e2eb603c6e2187362d515f3a32a7cb3c8a4b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:49:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb9bfd4d73f448e9686e02b928f4f03 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861592c18aaa4046a84f2a3f995da8e3 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41beacc794154ba6b37b8d6c5cbbc835 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e2de14892042bba658d713832e590c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce4c7fa2236946b2a87e50bb75359883 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c5231a7da549b6a8d4c3b940d5dc11 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff654786df7423ca364b6ff2cd09f2e : SUCCESS in 4m 39s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24ee1e5ae71a4e5ca7dd111106ab2131 : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0e436ad850d947e29fff143f4a626919 : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0864541b50716c7348974b37780e4ccab81e1d31","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-12 19:47:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cab723399ff42e3bca29e400b6072620a131d916","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 19:47:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c019c46fede2afa1b6568f35f19534af00b6e8bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:22:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0652edde4a8f4e9588aa84dd91fd09b8 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74fbdf5bc8ed4013a3a3de58cff203fa : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82831b8b1f11482fae9c7c0912d9177b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fde1c11e3ab423da1d57b1c0cd162b3 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc06d7e13dfb47938fcd666e4ff6b1bf : SUCCESS in 5m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cece306a055449e3a3da9fc24397654d : SUCCESS in 1h 33m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/72750a7249b8461aa90eb9c61e607f13 : FAILURE in 51m 09s","accounts_in_message":[],"_revision_number":2},{"id":"02764d7342c6d1b301406f65e5996b69b391a4b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 06:26:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"24170c57087f91bb74b1eb97f1ce39709c539e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 07:59:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a436b82423849adbec99d7975e56f98 : SUCCESS in 11m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc2e07a7f0d64c38926cda1790d58db6 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c2083bbb2064108b8d5fdcf593128a9 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a0b53f922c4af9a17d4b575d80af0b : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f5a028039e2497f9bc84b5ceb0b5095 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7c8c11d2404068840f36f3c5e480af : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aea8b5bf512c4fb99b2920d7365cc27d : SUCCESS in 7m 24s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0148430d8f124799955a3256adffaf06 : SUCCESS in 1h 30m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b67fd0e23c0b406db38160381b430483 : FAILURE in 45m 14s","accounts_in_message":[],"_revision_number":2},{"id":"688a0261a3ef545fcbaaefd380f7d91217ac68cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 11:21:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2be458c9aad7ef181712e27ecd0298237be3204b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:31:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd93fdef4ffa41ea942e1e3563d2985a : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e44ea50cd6745048902c944b6ec3cd6 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0d7344b30024b2e9ef1f8dc1668fc90 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b91a74e1d244a9b967631efae183085 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2c4de84886436a8d479fb0bee80905 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc77facd218e44fbb07f659d52aefe6c : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ca7b45944e14aeca2cb3e9c97229164 : SUCCESS in 5m 28s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66ecb04e778f4e4ba718e65042286368 : SUCCESS in 1h 05m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d90d50da8ec34d0fb8ba7e796ca05203 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0f67bcea17937dcce5fc036569a16e710c1fd7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:31:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6eec83dbadf257093bae7548c5b012186dbcb690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:59:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a6f2f19a654ff2a9d715d967a99506 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2ab6c55f29c477f8ffca2e9b28436fa : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31841c13befa442cb78f8585601b8832 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/598fadacf0cc4d75b19161bdd61956fb : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fd4d60b362147d9b2d96c59f59adab1 : SUCCESS in 3m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61bff78f53b547809df1892ddc795bf5 : SUCCESS in 1h 18m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a1a4f705dc624c87b2a7e0ada66cfa24 : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0782ddba2f94e68be3f87e13b84adf9849718f33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:59:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9b6d226b8a8b9e1e31d9d71ebdba150a1e469b01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:00:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a8d0af6477441d93d3549aeb343acc : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dc91446d39372974559c4ce66346549f0a288954","revisions":{"f3c8eb0f539332d6a36031f398b782b8096a6c27":{"kind":"REWORK","_number":1,"created":"2021-05-07 17:27:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/790285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/790285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/1"}}},"commit":{"parents":[{"commit":"9d3cac1d69a143c0a44230144760dd86c52d6ac7","subject":"Modify \"port.mac_address\" API definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9d3cac1d69a143c0a44230144760dd86c52d6ac7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 17:25:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 17:27:13.000000000","tz":0},"subject":"Add \"cache_ok\u003dFalse\" flag to TypeDecorator TruncatedDateTime()","message":"Add \"cache_ok\u003dFalse\" flag to TypeDecorator TruncatedDateTime()\n\nChange-Id: I33f42ec487567327c7c96afd81dd61e9e091c08f\nRelated-Bug: #1927763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f3c8eb0f539332d6a36031f398b782b8096a6c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f3c8eb0f539332d6a36031f398b782b8096a6c27"}]},"branch":"refs/heads/master"},"dc91446d39372974559c4ce66346549f0a288954":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-11 08:19:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/790285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/85/790285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/85/790285/2"}}},"commit":{"parents":[{"commit":"9d3cac1d69a143c0a44230144760dd86c52d6ac7","subject":"Modify \"port.mac_address\" API definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9d3cac1d69a143c0a44230144760dd86c52d6ac7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-07 17:25:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-11 08:19:05.000000000","tz":0},"subject":"Add \"cache_ok\u003dFalse\" flag to TypeDecorator TruncatedDateTime()","message":"Add \"cache_ok\u003dFalse\" flag to TypeDecorator TruncatedDateTime()\n\nChange-Id: I33f42ec487567327c7c96afd81dd61e9e091c08f\nPartial-Bug: #1927763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dc91446d39372974559c4ce66346549f0a288954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dc91446d39372974559c4ce66346549f0a288954"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
