)]}'
{"id":"openstack%2Fneutron-lib~791978","triplet_id":"openstack%2Fneutron-lib~master~If554c599bfcca694d3db15738d581b57be6ab04d","project":"openstack/neutron-lib","branch":"master","topic":"remove_context_session_override","hashtags":[],"change_id":"If554c599bfcca694d3db15738d581b57be6ab04d","subject":"[DNM] Remove warning about session with and without enginefacade","status":"ABANDONED","created":"2021-05-18 14:03:11.000000000","updated":"2021-08-15 08:29:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d0187b7ce623d2bdbe2f695b254edcdef8d18adc","_number":791978,"virtual_id_number":791978,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-22 09:12:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-05-18 14:03:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-18 14:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04c42be82582e68cd07dd864e86aba494eeba26b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 14:03:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dccc390c025325207b8f65c0f89c314c050d8234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:38:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202cfa47f7074ed78a4e008f6aabb75c : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a6b3804fa74445ebd6a190139d85359 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b067d14e6824b91bbdaadde9aa7bccc : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7a8971c1fc40129d781c0d201b2456 : FAILURE in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceebd33c635143a8865ddaba5af15bd6 : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a393b617108d428087117e0af40ad9fd : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4a43f783e9448d7bee60df7ba371f27 : FAILURE in 5m 31s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edaf745170ff42cea99084e1e73fc337 : FAILURE in 28m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f3cfeb6c721a432c93626cf79cb2edaa : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"19b30988ea2263c7fa1ff76067ad09180d83a017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 15:28:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ffbd6eb69886a8303f487263359fdf441a040a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 15:59:37.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4aa163bc624f64810dd5cfc10b14ab : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e805e75ffb8846aca5734c50f2225246 : FAILURE in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c11dd881c4b4484290444d81a7e0af97 : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c7d3b0dd74465b8bf079879c75e468 : FAILURE in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810792d89080430695e662b39e024952 : FAILURE in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed6161f5ac304950b806a7901ad8a6aa : FAILURE in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e2cb0c8b0e45baac2e7b378a66a501 : FAILURE in 4m 17s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c15e1139cb3e46889d76743355e0cc1c : FAILURE in 26m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca19bf9f2ba3416683b8eba21e9faaba : FAILURE in 23m 10s","accounts_in_message":[],"_revision_number":2},{"id":"eefbe108551f86ce0318626a33379a9fe1b53992","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-21 13:21:44.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5e6246943ca9916b66fc4f54819c526ced105bee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 11:13:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f8ffe183e8a11cfa7664bb5f3ad773c28e3333b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 11:14:03.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"aa2fac3a587a2bae18f3d1f2304cdae2e3e6b750","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 11:14:09.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"0278cbf22daee6a9bf51748404cf150a0079de95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 11:48:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38501f7478ae4b44aaa90d4f65c96876 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11e0473604841abae660191aeb3116d : FAILURE in 4m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6a9929543614927b9ef3f06442aeb8a : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31c44d8a31340cb84a931f73c0fc3bc : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2b1773505c49008ac49a88848db1e8 : FAILURE in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce0cbf6af085419185fe50d1d1d71783 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fab445d011864fd8b025cc115a0a3fac : FAILURE in 4m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee279a0a83d04e54bcaa2e4e2f124389 : FAILURE in 28m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/758315b465754fa09ce7b13b904d29fc : FAILURE in 26m 43s","accounts_in_message":[],"_revision_number":3},{"id":"5ee83cee99f78a8899e15a729420dbfc33b1414b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 08:45:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1ace331d53a4034397b11ccffc2557df279a2235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:12:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f197457e15c4b56b0e85655ee40e0c4 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e380c876772450e90c306e6c01a29cc : FAILURE in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/348d3db1f6154034a2b3eee7bdcf8d33 : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca2f12654c64764a79d72999d27d905 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbd22bce49434de69707cdf4ec1a9e18 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d66e6bc4fef4498ad8f874f6bc0198a : FAILURE in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94f9bf624c2e4655a55ddb0a319b873a : FAILURE in 5m 20s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c492d5350fd437998253af7211111fd : FAILURE in 22m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/defb13fa39b14991b7079c46b9cd5201 : FAILURE in 19m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d0187b7ce623d2bdbe2f695b254edcdef8d18adc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-15 08:29:45.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7092aae9368f6e1f39fa36ab5d18c93d35e8afc0","revisions":{"09f6a8fcb60660934f2a1f6f56fb1b2ba55c69db":{"kind":"REWORK","_number":1,"created":"2021-05-18 14:03:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/791978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/791978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/1"}}},"commit":{"parents":[{"commit":"8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a","subject":"Merge \"Move bgp api-def to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 14:01:07.000000000","tz":0},"subject":"[DNM] Remove warning about session with and without enginefacade","message":"[DNM] Remove warning about session with and without enginefacade\n\nTesting patch\n\nWith this patch I want to find all places in Neutron code where a\ncontext is used without an explicit transaction declaration. We\nshould manually add them to Neutron.\n\nChange-Id: If554c599bfcca694d3db15738d581b57be6ab04d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/09f6a8fcb60660934f2a1f6f56fb1b2ba55c69db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/09f6a8fcb60660934f2a1f6f56fb1b2ba55c69db"}]},"branch":"refs/heads/master"},"afae8ebc2d36eb7611f01c7a8c508fc4508815df":{"kind":"REWORK","_number":2,"created":"2021-05-18 15:28:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/791978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/791978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/2"}}},"commit":{"parents":[{"commit":"8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a","subject":"Merge \"Move bgp api-def to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cfd261349ee9fbc8a657ce82eeac1f39b3e1e6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 15:28:31.000000000","tz":0},"subject":"[DNM] Remove warning about session with and without enginefacade","message":"[DNM] Remove warning about session with and without enginefacade\n\nTesting patch\n\nWith this patch I want to find all places in Neutron code where a\ncontext is used without an explicit transaction declaration. We\nshould manually add them to Neutron.\n\nChange-Id: If554c599bfcca694d3db15738d581b57be6ab04d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/afae8ebc2d36eb7611f01c7a8c508fc4508815df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/afae8ebc2d36eb7611f01c7a8c508fc4508815df"}]},"branch":"refs/heads/master"},"199489d8e8473dfa25e166c5ee2aa4a807bdaa49":{"kind":"REWORK","_number":3,"created":"2021-06-04 11:13:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/791978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/791978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/3"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-04 11:13:06.000000000","tz":0},"subject":"[DNM] Remove warning about session with and without enginefacade","message":"[DNM] Remove warning about session with and without enginefacade\n\nTesting patch\n\nWith this patch I want to find all places in Neutron code where a\ncontext is used without an explicit transaction declaration. We\nshould manually add them to Neutron.\n\nChange-Id: If554c599bfcca694d3db15738d581b57be6ab04d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/199489d8e8473dfa25e166c5ee2aa4a807bdaa49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/199489d8e8473dfa25e166c5ee2aa4a807bdaa49"}]},"branch":"refs/heads/master"},"7092aae9368f6e1f39fa36ab5d18c93d35e8afc0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-22 08:45:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/791978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/78/791978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/78/791978/4"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-10 11:56:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-06-22 08:45:55.000000000","tz":0},"subject":"[DNM] Remove warning about session with and without enginefacade","message":"[DNM] Remove warning about session with and without enginefacade\n\nTesting patch\n\nWith this patch I want to find all places in Neutron code where a\ncontext is used without an explicit transaction declaration. We\nshould manually add them to Neutron.\n\nChange-Id: If554c599bfcca694d3db15738d581b57be6ab04d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7092aae9368f6e1f39fa36ab5d18c93d35e8afc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7092aae9368f6e1f39fa36ab5d18c93d35e8afc0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
