)]}'
{"id":"openstack%2Fneutron-lib~795423","triplet_id":"openstack%2Fneutron-lib~master~Ibee622ef47ca6d738ca53f6482cad88b2441503e","project":"openstack/neutron-lib","branch":"master","topic":"rbac-bgpvpn","hashtags":[],"change_id":"Ibee622ef47ca6d738ca53f6482cad88b2441503e","subject":"Introduce rbac-bgpvpn api extension","status":"MERGED","created":"2021-06-08 21:56:25.000000000","updated":"2026-09-14 14:25:15.000000000","submitted":"2021-07-06 09:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"795423-rbac-bgpvpn","meta_rev_id":"dea8bfd07800f87508c6b205e67a693129a06909","_number":795423,"virtual_id_number":795423,"owner":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 09:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-06 07:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-07-06 06:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":2,"date":"2021-07-06 07:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":1,"date":"2021-07-06 07:52:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"}],"CC":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 23:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 07:26:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-19 19:13:05.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2021-06-22 13:05:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-29 09:02:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"39de3ff91bf85f3854d71161bc1c9931229028d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-08 21:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00a96bbe29fe31c69b1e515fa49f3c8bbcf5606f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 23:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7bfeb645053a49f59a43b4f3a1f71222 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770948e0c7ab43c3bbf62403e1377540 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e5509b47b6745168323f131cd88c707 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b58f53a3a1c4f6ba5b707b0df9e500f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944b035dab884c78b0618e8398923189 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9b4fe7abadd4ab7b5553643ccbb14a1 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dd1ce9ef273491e90f7cd0b3c91fa4f : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e7401171e24cf4bcf9a003ff65c8fc : SUCCESS in 4m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/287bbc4c72dc4cefa63ec78af902c158 : SUCCESS in 1h 27m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1898a5c90fff490195f2628af64ffb7f : SUCCESS in 3m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4be85f826cc14b1bb29eb2b8c32012a2 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8effd584f5331ce81d82bc0b07ac721c4c25dd82","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-09 07:26:49.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is just \"procedural\" -1 for now. As this is RFE, lets first discuss it on the drivers meeting and then follow up with the implementation (if it will be approved of course)","accounts_in_message":[],"_revision_number":1},{"id":"bbf52b6825859deb556144d5cb0dee10303f7398","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-19 19:13:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07b98c936986caa2d55404d9b0ebaf7e012a71ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-22 12:48:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d4f58bab46bd04ea9ecd17d4862e6952db5f2ec","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-22 12:49:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"4ce093f3f33c2eb84c2f186886b8fbb139f95067","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 13:05:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3711670735047f40ecbbd26ba5808a986ba8028e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 14:02:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f3c824ec8416416eb4dfa3c4b5f44ece : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f0b2ffe0ad49849cd304564b89e3bb : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aab6da979c154938afa3f0fa00a00884 : FAILURE in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0020af3cf77e4c77bb4de87693d9e749 : FAILURE in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b820fcde48b24cef918bf9f0efef3bd9 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8091e5e2cf7423e8c23454357c7b0de : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a46a02c77e9f4a34a793248cf3e2de9c : FAILURE in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaa4af533f47418b8fba11bd3321a715 : FAILURE in 5m 21s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ed9473844e7477798b30220f3b4c052 : SUCCESS in 1h 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8f0e2bd75844e749e07a3af54bec4ac : SUCCESS in 3m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1b88a991379440ef8184e70adf52264d : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c25ed58b8836d2e6b5aa5440cf82ebc433456894","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-25 19:07:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2e425b8692486fe9b5eb32c244ff28dc75efc57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 20:33:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/25745e536bf441d6af836900422ad3c9 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f1e6acd8584892936910ebaa5b474d : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cba5572ebdc4c84a1b2165e10b15be1 : FAILURE in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1796ecb3c8204b12b532275299b72254 : FAILURE in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852f2bae5d74451192986231ccc1a00d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7c38a2b1cf4a199979ba55c6dd0146 : FAILURE in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f270be8aaa0248bdbc872e24c7fea59e : FAILURE in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5abfb95c3ba4e8ba7580c77dfa93599 : FAILURE in 6m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40cd317238684368b5343010e4ad4ca9 : SUCCESS in 1h 21m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6170e7d81afc4c04bef0cc6f43834280 : SUCCESS in 6m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e30eda92cad4a929cdd2b0280be2594 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5e1b180969831353fdb28fe6d7656fe984a5fc1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-27 13:02:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"226ece21816e72765c32858700150339b62f0dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 14:29:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fbec737f1b704d2298bb41c73b6e119e : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb519ce4a0a4aa5990b6c47d9644730 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75731e7ca584ac7adeefedb48dead1a : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44ad49af353442458371255893a69156 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ab204a1a274b7ab3e04bb3fbacc93d : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f648121e6ec46ceb335a6607a0fa74c : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/243e570b0f5943b4a5622956ec90f3f8 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06521f9d3b484a0d86a7d922c00524b7 : SUCCESS in 6m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7775ce669d964ba6887c34722fa12915 : SUCCESS in 1h 25m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9540e42e8c2945b49f3c193e3e10be42 : SUCCESS in 5m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb3627a5e1724fe3a2bc1afe131fbb0a : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":4},{"id":"c8779ccfa3f4a2fe04ad47286a6b58171811c0dc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-29 09:02:41.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nThese are more questions, overall looks ok","accounts_in_message":[],"_revision_number":4},{"id":"d0f3d55487b9958904fef4ad278574cc4a09253b","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-29 21:29:53.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0487f9c24c4844a5df2d14345885d275f86c343a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-29 21:31:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"79b5e53b584a84942a1987f618ef9f51724a2e96","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-06-29 21:41:47.000000000","message":"Topic set to rbac-bgpvpn","accounts_in_message":[],"_revision_number":5},{"id":"9a4f934afb6d1bc027634c17295da43460a35638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 22:49:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0061c1bcf3147ec9804c84c4e7b090e : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f430ea4613e4539825f3ab570c11431 : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de5fc9f4594846d88b917b8818a762a6 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e27fe077794e048b4bc43dafad4b71 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02314f9c31444378bb266f3a7ad8c1b1 : FAILURE in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5eb6ce8e53d4e4d83d8936db772c2a0 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0f080e2f0204659ada2c7eb072780cd : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b5a2f7043c4fc0b4ea57d6bdd53e6f : SUCCESS in 6m 50s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60f74ea1d0d647e8a09aedc10635a37f : SUCCESS in 1h 15m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e765bcc494485c8d92c02aadb3d6cd : SUCCESS in 5m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3c62400e19dd45c39b5610a06a299992 : SUCCESS in 43m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fa2e5b1e50c45e19e6a297479954e13 : SUCCESS in 46m 55s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/extensions.py","accounts_in_message":[],"_revision_number":5},{"id":"d5c04c3cab1d770865972140e5c9725dda6b9431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-07-01 10:20:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5701bbd1c30881bd159d75b27a637f21f8287cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:43:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eb47cb8308a547d4955df2919fb24963 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2cdf79a194540408150a3425dce208a : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b93c2a192094ca8b593a8edaa38a2ff : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acc81523c5aa4dbfa6a129e75d9bcccb : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df223e937b6741ee85256ddb0663c42d : FAILURE in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c84002e3871491b930b74249f1f6c7d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ded26d174e24a2ea63bb1f3032d01b8 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee30309344444e0f9667cc023a086dbb : SUCCESS in 4m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39a1e3b2743a4fd8bf5c4d9a68c1d58b : SUCCESS in 1h 21m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/428b78ed83f54cf98126024293dd1c96 : SUCCESS in 4m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71361019a3144b82aa600aa236e6dbe3 : SUCCESS in 39m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bc081ee965840358f1eeab56f141ae5 : SUCCESS in 45m 00s\n\nWarning:\n  Comments left for invalid file neutron_lib/api/extensions.py","accounts_in_message":[],"_revision_number":6},{"id":"cb9c7a4283c9578ec525198d924a17628d6dddb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-07-05 18:30:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"852a2b803981e5974490a787ef623a3bae432580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 19:44:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6beb1632f047492ea618b70317c0d713 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1fa2f99002499f8361c61ec7afb7b3 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c49426edc0504d3891d1a60f610c7148 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c755f68e59e144a2a9b862a000130a92 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f52b2b5318441fa1a56161305d170b : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2783930cca451790058e28b3d90ea9 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4f9953aad764589900bf561b2c3c14d : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61821c4b7fae465a9ddc7f446784385b : SUCCESS in 4m 03s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99ae61c000324219ab71ca13c7de533a : SUCCESS in 1h 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/132a02a684fa4b2c9227e9d3a817a0a1 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c80c8e3ed390474fa46a4a2bff5dbd0b : SUCCESS in 34m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75c59ece80ef449e8eebf3469cc64dd2 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"430cf86bb9af972b8b219ef39b0ffed0a84ef8be","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-06 06:14:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"aae730f23b8b8c2d84db75e6e8d1610bba28b190","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-06 07:49:01.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0511b446a218d3e578942935c869720784829dc2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-06 07:52:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nAll things/questions answered","accounts_in_message":[],"_revision_number":7},{"id":"539934b99b280db72bb42b347cdc626b994f027b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 07:53:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"33cb2bbe3c01f9ebcb3c3e542e9b07787cd9b841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 09:10:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9caecea5a424814a4d77e019fdf9486 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736db0126fbf4c1ab569fb66fade5d7f : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79f62386d19b49c6b3bd2ecff745ef05 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5d29da301c46ff9443bdba81657f0d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9722a7d117ce43ec8deee9a60b264a6a : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c082bad04c5411a8624937de73c7262 : SUCCESS in 4m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7358a3ceb2f455fad51e10855cd4470 : SUCCESS in 1h 16m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfcc79baa75b48e39b9da22cba3d6f4f : SUCCESS in 3m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a10d32f508fa471eaa6faa7db65cc190 : SUCCESS in 37m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/494ffe5fa99e496a9266f583b2ed76c6 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":7},{"id":"e2d3a30986ab3ca78a359dbf517c46e592693b09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 09:10:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b04cfcbeeebbbbac5dcf8c013b75489598d36d39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 09:12:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/fc4801b6d52846a09b8b25180bc9d2e0 : SUCCESS in 1m 21s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f06510e8a5e4aa9a4fa4759c21fb984 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0515a7c5c03c44ecbd857a1d3d431a34 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7},{"id":"dea8bfd07800f87508c6b205e67a693129a06909","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-14 14:25:15.000000000","message":"Created a revert of this change as I02f7bba1f897a0d60063da1bd778944d7f5a5ca3","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2cfe12c2ece452882b0646036fc01fd021cf727a","revisions":{"06df0ddb0e0fa4a5998d961eee15ae8c9831b404":{"kind":"REWORK","_number":1,"created":"2021-06-08 21:56:25.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/1"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:55:58.000000000","tz":180},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06df0ddb0e0fa4a5998d961eee15ae8c9831b404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06df0ddb0e0fa4a5998d961eee15ae8c9831b404"}]},"branch":"refs/heads/master"},"143c0ff140efd96c6e373cd12caea91c579041fe":{"kind":"REWORK","_number":2,"created":"2021-06-22 12:48:46.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/2"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-22 12:40:14.000000000","tz":180},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/143c0ff140efd96c6e373cd12caea91c579041fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/143c0ff140efd96c6e373cd12caea91c579041fe"}]},"branch":"refs/heads/master"},"03b8e4a7f0b7c6cb9ef876f7f444721dd96eb8b7":{"kind":"REWORK","_number":3,"created":"2021-06-25 19:07:46.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/3"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-25 19:07:22.000000000","tz":180},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03b8e4a7f0b7c6cb9ef876f7f444721dd96eb8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03b8e4a7f0b7c6cb9ef876f7f444721dd96eb8b7"}]},"branch":"refs/heads/master"},"e9de8ab5b748c0843c2d276e6815614efb05b65a":{"kind":"REWORK","_number":4,"created":"2021-06-27 13:02:47.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/4"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-27 13:02:28.000000000","tz":180},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9de8ab5b748c0843c2d276e6815614efb05b65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9de8ab5b748c0843c2d276e6815614efb05b65a"}]},"branch":"refs/heads/master"},"2b36a9ab5f18be40028901eaea56cab19ea9bee0":{"kind":"REWORK","_number":5,"created":"2021-06-29 21:31:49.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/5"}}},"commit":{"parents":[{"commit":"9ae481ef9e9654caccf7664be66bc7374af750b6","subject":"Merge \"[api-ref] Add HTTP 501 for PUT min bw QoS rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9ae481ef9e9654caccf7664be66bc7374af750b6"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-29 21:31:30.000000000","tz":180},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b36a9ab5f18be40028901eaea56cab19ea9bee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b36a9ab5f18be40028901eaea56cab19ea9bee0"}]},"branch":"refs/heads/master"},"60da76e6ec0252549723e636a236ec98a4546fa6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-01 10:20:36.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/6"}}},"commit":{"parents":[{"commit":"63c8443a65db4c662237b8687beff19c3973b819","subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/63c8443a65db4c662237b8687beff19c3973b819"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-07-01 10:20:36.000000000","tz":0},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60da76e6ec0252549723e636a236ec98a4546fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60da76e6ec0252549723e636a236ec98a4546fa6"}]},"branch":"refs/heads/master","description":"Rebase"},"2cfe12c2ece452882b0646036fc01fd021cf727a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-05 18:30:44.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/23/795423/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/795423/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/795423/7"}}},"commit":{"parents":[{"commit":"66d203d6275ed987362d117d4edc30b48b291207","subject":"Merge \"Rehome DHCP_IPV{4,6}_TABLE constants to neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/66d203d6275ed987362d117d4edc30b48b291207"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-06-08 21:46:41.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2021-07-05 18:30:44.000000000","tz":0},"subject":"Introduce rbac-bgpvpn api extension","message":"Introduce rbac-bgpvpn api extension\n\nThis extension makes it possible to add bgpvpn to RBAC policies.\n\nPartial-Bug: #1931100\nChange-Id: Ibee622ef47ca6d738ca53f6482cad88b2441503e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2cfe12c2ece452882b0646036fc01fd021cf727a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2cfe12c2ece452882b0646036fc01fd021cf727a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
