)]}'
{"id":"openstack%2Fneutron-lib~797281","triplet_id":"openstack%2Fneutron-lib~master~I0ea6cf4cf9a207863246a3669bc4001293f96b87","project":"openstack/neutron-lib","branch":"master","topic":"bug/1932483","hashtags":[],"change_id":"I0ea6cf4cf9a207863246a3669bc4001293f96b87","subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","status":"MERGED","created":"2021-06-21 10:54:36.000000000","updated":"2021-06-29 09:33:47.000000000","submitted":"2021-06-29 09:31:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797281-bug/1932483","meta_rev_id":"1a9dcdb1c30266708403b3f4e3064cb0f8b33169","_number":797281,"virtual_id_number":797281,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 09:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-06-24 07:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-06-29 07:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 07:54:20.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 07:54:20.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-21 11:05:19.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-06-21 13:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 14:19:13.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-06-22 15:00:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-22 15:26:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"97616c7d7b5fcf59b7fea42f8379522a4a68de96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-21 10:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da69f78ed2b972e43461f7f261d5536c073be9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 13:41:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d333d053754df1a5e7e3ad90c45414 : SUCCESS in 17m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d8f18de7cf54c1e9b06f81df35750ac : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/216c8a59e62c451f8a0444a4ae16dd35 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92475be4fcaa4897bb3a2444285a6d7c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2caecb2638e041e8977f4fe4bba3c39c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfdd7c8fe0c24a178a8d959ca86df1c4 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d24c7d51b9c4ba1a3c30971329690bd : SUCCESS in 5m 09s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2968976aa42041edb76c76e293ea3a27 : FAILURE in 2h 40m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8e7565cf4b12411f92aea1129b4dbac9 : SUCCESS in 1h 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff6eff37eb104e4ca1d5b3150a287864 : FAILURE in 49m 25s","accounts_in_message":[],"_revision_number":1},{"id":"94dbd4a23b32cda5500efb42f1e0b11c92016ca9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-22 07:21:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"207df427fa74b3d6c46c03fa7078e2861806c573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 08:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af28651a54a745598564037a3dab5427 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/742c6461526c49898da76849f2c74293 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef51f4d59993436e8ab5bb7781b212dd : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa604bd0dd0f4eb0bff000d510f8b004 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07452011db2a4944826a13b21ea01b00 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f930f36291c4504bed0f0e49cecd593 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fa2219f796d406e8fca7e6edb6173d9 : SUCCESS in 4m 34s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64657d22a207473db2a53b2ea3540114 : SUCCESS in 1h 20m 18s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42de65c0daa143c798aa8991755a5fdf : SUCCESS in 34m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6aaf9a74907481a92acc87e264ccd9f : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8e6f409c53d1e02def40c44828af27ba33e39049","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-22 08:54:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6efa046bfd2ddbd7c2823620bbc8df4cfe3958e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-22 14:19:13.000000000","message":"Patch Set 1: Code-Review-1\n\nLooking at the log of neutron-functional-with-uwsgi, the release version of neutron-lib is used.\n\nhttps://zuul.opendev.org/t/openstack/build/c6aaf9a74907481a92acc87e264ccd9f/log/controller/logs/pip3-freeze.txt#52","accounts_in_message":[],"_revision_number":1},{"id":"a2558eb2366ab4b0989656fba78d8e1ada2ed385","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-22 14:49:44.000000000","message":"Patch Set 1:\n\nMy previous comment looks incorrect. This is a functional job, so devstack is used only to install\nrequired ubuntu packages and pip-freeze output in the devstack log is not the right place to check.\n\nOn the other hand, neutron-lib from PyPI is still used.\nLooking at the log of tox -e functional in job-output.txt, we still see the released version of neutron [1].\n\n[1] https://zuul.opendev.org/t/openstack/build/c6aaf9a74907481a92acc87e264ccd9f/log/job-output.txt#11810-11811","accounts_in_message":[],"_revision_number":1},{"id":"9b235dc96aa0846ff33fde32f6c2080e37355dbd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-22 14:55:05.000000000","message":"Patch Set 1:\n\nthe reason that neutron-lib is not installed from the git repo is that \"tox_install_siblings\" is set to false in neutron-functional job. I am not sure why it is set to false and wonder if there is any side effect when we set it to true.\nhttps://opendev.org/openstack/neutron/src/branch/master/zuul.d/base.yaml#L37","accounts_in_message":[],"_revision_number":1},{"id":"4d02d6633acae1f371c6cca4eb9d580037b0d101","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-22 15:22:15.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"a7bfad7c427ecacb05c52fbd8e5749ef12c5c556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 11:13:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1464796394262c199e12f127f04904b9d86d6d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 12:32:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fd179ddd5143448d70bd61cf2b031b : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2e2a38acb04c938a06c23efce03118 : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4538ea9a985b4000a6a9af51fb03fb3f : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23454a462404497da0aea221897e9fe2 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19ea8a97448439a8c6c4f98114919c6 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b5744bd6fec44e384badceb8112a1c9 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27d9138a145d468bac633f499719cc00 : SUCCESS in 4m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/caf653139b1f4534b780a1da5f7f942c : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/26786b600a7e48ba95f43d0fa7c5f561 : SUCCESS in 46m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00e5accf777d4b6692beaefae897636d : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"16e05f5c8ce0114cd8ded76c2596f68972568b03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 13:42:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa5cc5ba6e8c4d935514c615589832ad15914713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 13:44:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c299ac9182b3ee1562e41a397d6a69b403bf26ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 13:45:12.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-lib on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027neutron-functional-with-uwsgi\u0027][\u0027required_projects\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-xena-jobs\n        - lib-forward-testing-python3\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-tempest-plugin-api\n          - neutron-functional-with-uwsgi:\n              required_projects:\n                - openstack/neutron-lib\n          # This job comes from lib-forward-testing-python3 template,\n          # but it is limited to 2h there and we want to set\n          # for it 3h timeout\n          - tempest-full-py3:\n              timeout: 10800\n      gate:\n        jobs:\n          - neutron-tempest-plugin-api\n          - neutron-functional-with-uwsgi\n          # This job comes from lib-forward-testing-python3 template,\n          # but it is limited to 2h there and we want to set\n          # for it 3h timeout\n          - tempest-full-py3:\n              timeout: 10800\n\n  in \"openstack/neutron-lib/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":3},{"id":"0cac756064ce07f8925db1a8e2c918d3f05876e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 13:52:04.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"86496cd7d984e664b6a844cdd5e7f0f09560f9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:40:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd6ea9e93a645769d6ae15540a2ae06 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3332369cfc84456b8ef1d82012e8d215 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01ac8d8a1cd642ec8deed1f3c936b94c : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4287a3a21e4f028bcc9634b6bde6b1 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31ed822a823e42619001b5ece71f1cda : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f91e69bdcc4639bbe944b9ef03256e : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eac94e0610047f39bff94f577dbd7d6 : SUCCESS in 7m 18s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/751cbf9160c04a4192c88f2e2ce8233a : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8bccaccc69f14c7c859e17af443d9398 : SUCCESS in 42m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a58192740b145a8922ab5f4ba856690 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":4},{"id":"89cc0831301bab943c2fabd76de753abeed07ac8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-24 07:35:45.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good and passing with tox_install_siblings\u003dFalse - and I see neutron-lib @ file:///home/zuul/src/opendev.org/openstack/neutron-lib in pip freeze","accounts_in_message":[],"_revision_number":4},{"id":"894c9865c82b439ad95ea4a43d83ad9923df59b9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-29 07:48:06.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e Looks good and passing with tox_install_siblings\u003dFalse - and I see neutron-lib @ file:///home/zuul/src/opendev.org/openstack/neutron-lib in pip freeze\n\nWhat we need to check here is which neutron-lib is used in the functional tox env rather than pip3-freeze.txt in the controller log (as the functional job uses devstack to install our dependencies of ubuntu packages).\nI confirmed neutron-lib master is installed into the functional tox env as expected. [1]\n\nhttps://zuul.opendev.org/t/openstack/build/5a58192740b145a8922ab5f4ba856690/log/job-output.txt#12029-12143","accounts_in_message":[],"_revision_number":4},{"id":"2cf55262cf42938f134851917accf509ac3f7ddf","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-29 07:54:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cfc8b674f3c548f821f308a4cd54a35e3a997e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 07:54:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6b3d53f7b57e7510eacc34bc2d9f499ad46c7c12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:31:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd1f642c3f44d77883cdbc6382c875e : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab162a00f5154e3ba565082973daffe3 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e5472a476547718866284d378ecf72 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03e1015ddacc4799af9d909da87884ad : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ee706172c8e492fad6de19af5e38b52 : SUCCESS in 5m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0e2fcb0af784df4b300ff4a1a7975f5 : SUCCESS in 1h 31m 25s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5042ac873624d5899f179bbb43fc4f1 : SUCCESS in 45m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c50673abb734c248089027402784ef1 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a4925f4eee35243b904f580dd240fb5135e1899d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:31:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1a9dcdb1c30266708403b3f4e3064cb0f8b33169","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:33:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd747ce661f48c6bac985ac389683c3 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"63c8443a65db4c662237b8687beff19c3973b819","revisions":{"d0e339cb96c534730ecb7dc727d48434ca7b9a3e":{"kind":"REWORK","_number":1,"created":"2021-06-21 10:54:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/81/797281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/797281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/1"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-21 10:52:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-21 10:54:32.000000000","tz":120},"subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","message":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI\n\nThis job can help find potential issues early and that may help\nto avoid Neutron\u0027s CI breakage after new neutron-lib version is\nreleased. See related bug for example of failure which this could avoid.\n\nRelated-Bug: #1932483\nChange-Id: I0ea6cf4cf9a207863246a3669bc4001293f96b87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d0e339cb96c534730ecb7dc727d48434ca7b9a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d0e339cb96c534730ecb7dc727d48434ca7b9a3e"}]},"branch":"refs/heads/master"},"1db6e94387dc5c163da5acb766ea32e9f5f1fcfe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-23 11:13:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/81/797281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/797281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/2"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-21 10:52:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 11:13:06.000000000","tz":0},"subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","message":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI\n\nThis job can help find potential issues early and that may help\nto avoid Neutron\u0027s CI breakage after new neutron-lib version is\nreleased. See related bug for example of failure which this could avoid.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/797622\n\nRelated-Bug: #1932483\nChange-Id: I0ea6cf4cf9a207863246a3669bc4001293f96b87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1db6e94387dc5c163da5acb766ea32e9f5f1fcfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1db6e94387dc5c163da5acb766ea32e9f5f1fcfe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"890b814f2d7b9251d696b5fafa147c30bac35ea2":{"kind":"REWORK","_number":3,"created":"2021-06-23 13:44:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/81/797281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/797281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/3"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-21 10:52:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 13:44:16.000000000","tz":0},"subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","message":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI\n\nThis job can help find potential issues early and that may help\nto avoid Neutron\u0027s CI breakage after new neutron-lib version is\nreleased. See related bug for example of failure which this could avoid.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/797622\n\nRelated-Bug: #1932483\nChange-Id: I0ea6cf4cf9a207863246a3669bc4001293f96b87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/890b814f2d7b9251d696b5fafa147c30bac35ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/890b814f2d7b9251d696b5fafa147c30bac35ea2"}]},"branch":"refs/heads/master"},"63c8443a65db4c662237b8687beff19c3973b819":{"kind":"REWORK","_number":4,"created":"2021-06-23 13:52:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/81/797281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/81/797281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/81/797281/4"}}},"commit":{"parents":[{"commit":"c2879ec1739d1425bcc762d4b6584ee241f995ca","subject":"Merge \"Revert \"Add network parameter to some of the ML2 API methods\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c2879ec1739d1425bcc762d4b6584ee241f995ca"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-21 10:52:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 13:51:53.000000000","tz":0},"subject":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI","message":"Add Neutron\u0027s functional job to the neutron-lib\u0027s CI\n\nThis job can help find potential issues early and that may help\nto avoid Neutron\u0027s CI breakage after new neutron-lib version is\nreleased. See related bug for example of failure which this could avoid.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/797622\n\nRelated-Bug: #1932483\nChange-Id: I0ea6cf4cf9a207863246a3669bc4001293f96b87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/63c8443a65db4c662237b8687beff19c3973b819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/63c8443a65db4c662237b8687beff19c3973b819"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
