)]}'
{"id":"openstack%2Fneutron-lib~799445","triplet_id":"openstack%2Fneutron-lib~master~Ied4947ce0d52b84e501cac117f29882d6e6dba98","project":"openstack/neutron-lib","branch":"master","topic":"bp/address-groups-in-sg-rules","hashtags":[],"change_id":"Ied4947ce0d52b84e501cac117f29882d6e6dba98","subject":"[Docs] Add api-ref for the address groups","status":"MERGED","created":"2021-07-05 09:27:41.000000000","updated":"2021-08-12 09:11:46.000000000","submitted":"2021-08-12 09:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799445-bp/address-groups-in-sg-rules","meta_rev_id":"d2d155d3a1636265d745d49eb08927d3a0393220","_number":799445,"virtual_id_number":799445,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 09:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2021-08-11 21:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-08-12 08:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2021-08-12 08:14:33.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 09:28:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-05 09:28:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2021-07-05 09:28:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-05 10:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"325a93a575f3b71bbde3ce2f01a04e45aa6c7bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 09:27:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bab4829187cc3f907cbd1a94d30fc71b9c706eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 10:28:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d2a0d0659824b9f90d0cca00ec5a0dc : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77447edd9b214bfb98a9951a22a21814 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aca3d7baea04b1199504d9bdc1ba427 : FAILURE in 4m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c4323aeb12d48e2ac8dfc6453f53305 : SUCCESS in 41m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7b865f7ad794d7f8442e9cf4972bd97 : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":1},{"id":"39d98c38d0680d2f7de711f613c57fad15fe43e2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-07 06:05:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nWe cannot update \"addresses\" field in PUT request directly. It is read-only for PUT.\nWe need to use add_addresses and remove_addresses operations.\n\n- Needs to update PUT example to use \"name\" or \"description\" instead of \"addresses\"\n- Needs to document /v2.0/address-groups/\u003cid\u003e/add_addresses and /v2.0/address-groups/\u003cid\u003e/remove_addresses","accounts_in_message":[],"_revision_number":1},{"id":"912fcb8200e27a3f758f7860de7c4f687a72fbb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-07 06:34:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7991b1a95fa4acae21f47ea3a69ee032dc254a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 07:27:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3ead38c92d56487880312a9d36cbc207 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5924449bfe4249bea53f03cae55ce380 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a895730abf4a3eaa310301b231dc61 : FAILURE in 5m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6229a7c4ca8c4064a40c3d709b07dec6 : SUCCESS in 39m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/150df1976f7c4998a4530c86411a6161 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"581dfa23e3f2245917a5d09095bc1fbf4acf0498","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 13:01:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e We cannot update \"addresses\" field in PUT request directly. It is read-only for PUT.\n\u003e We need to use add_addresses and remove_addresses operations.\n\u003e \n\u003e - Needs to update PUT example to use \"name\" or \"description\" instead of \"addresses\"\n\u003e - Needs to document /v2.0/address-groups/\u003cid\u003e/add_addresses and /v2.0/address-groups/\u003cid\u003e/remove_addresses\n\nGood point. I somehow forgot about it. Now it\u0027s changed.","accounts_in_message":[],"_revision_number":2},{"id":"1ff101ad92a11384c6a3ff3883411722740dc94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 13:01:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a764fcd84ecefb2ff5c693379c1f1838bd5bd1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 14:04:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e2ebe3508674d5789adab34d12b43a7 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61568913c3f45c1a07633d080920b55 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b598eb828b874660be2df086dbc78d03 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df64d854803d474b916dd05f1bd1af6f : SUCCESS in 40m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22c461c340884cadae3758f80d45ed13 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":3},{"id":"790ff69a8f2443c3195fb0785a1cb00f0c6df4d2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-08-11 21:46:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f1498f263b33bb8f892e7356f9204fd7f7defc68","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-08-12 08:14:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1120675042edc09d3d860756b4af2ee68fd8cf63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 08:15:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c5830b20f4591fe26686de027e8d88485bc4f978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:10:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4793122731ca4d63b83c0a31c846252d : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56aab57ceb4f45089791d391548e0942 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7970ef1a4cf6413cb9eca294d49ae48c : SUCCESS in 4m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad89f723e86c4ff78a69cb2ea7278593 : SUCCESS in 45m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81873853297f43978a20d02f18014c63 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":3},{"id":"33fea9ab1d2f0fb2ce588523c3f729f7d9f93244","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:10:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d2d155d3a1636265d745d49eb08927d3a0393220","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:11:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ff5418ccdbc848d9abe08acfc9fe23a6 : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833ff4da576943e89f88e7d93d7f73a0 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3a07a644ef1461640a0c0fec3479b580f27d3eac","revisions":{"8174897f96fb05d0726d7aaf859b59e009d108dc":{"kind":"REWORK","_number":1,"created":"2021-07-05 09:27:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/799445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/799445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/1"}}},"commit":{"parents":[{"commit":"303258fbfc8be361c1ec7ab2c4d292336b490dfc","subject":"Fix pep8 check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/303258fbfc8be361c1ec7ab2c4d292336b490dfc"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 09:26:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 09:26:41.000000000","tz":120},"subject":"[Docs] Add api-ref for the address groups","message":"[Docs] Add api-ref for the address groups\n\nRelated-blueprint: bp/address-groups-in-sg-rules\nChange-Id: Ied4947ce0d52b84e501cac117f29882d6e6dba98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8174897f96fb05d0726d7aaf859b59e009d108dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8174897f96fb05d0726d7aaf859b59e009d108dc"}]},"branch":"refs/heads/master"},"f5030cb3bd3179dee61d5e42748ced2497f9bc9b":{"kind":"REWORK","_number":2,"created":"2021-07-07 06:34:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/799445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/799445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/2"}}},"commit":{"parents":[{"commit":"303258fbfc8be361c1ec7ab2c4d292336b490dfc","subject":"Fix pep8 check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/303258fbfc8be361c1ec7ab2c4d292336b490dfc"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 09:26:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-07 06:34:11.000000000","tz":0},"subject":"[Docs] Add api-ref for the address groups","message":"[Docs] Add api-ref for the address groups\n\nRelated-blueprint: bp/address-groups-in-sg-rules\nChange-Id: Ied4947ce0d52b84e501cac117f29882d6e6dba98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5030cb3bd3179dee61d5e42748ced2497f9bc9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5030cb3bd3179dee61d5e42748ced2497f9bc9b"}]},"branch":"refs/heads/master"},"3a07a644ef1461640a0c0fec3479b580f27d3eac":{"kind":"REWORK","_number":3,"created":"2021-07-20 13:01:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/799445/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/799445/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/799445/3"}}},"commit":{"parents":[{"commit":"303258fbfc8be361c1ec7ab2c4d292336b490dfc","subject":"Fix pep8 check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/303258fbfc8be361c1ec7ab2c4d292336b490dfc"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 09:26:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 13:01:31.000000000","tz":120},"subject":"[Docs] Add api-ref for the address groups","message":"[Docs] Add api-ref for the address groups\n\nRelated-blueprint: bp/address-groups-in-sg-rules\nChange-Id: Ied4947ce0d52b84e501cac117f29882d6e6dba98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a07a644ef1461640a0c0fec3479b580f27d3eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3a07a644ef1461640a0c0fec3479b580f27d3eac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
