)]}'
{"id":"openstack%2Fneutron-lib~800441","triplet_id":"openstack%2Fneutron-lib~master~I8b1302bbdb57523f98b07b3ca88778d3a3103050","project":"openstack/neutron-lib","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","hashtags":[],"change_id":"I8b1302bbdb57523f98b07b3ca88778d3a3103050","subject":"Add utils for parsing pkt processing cfg options","status":"MERGED","created":"2021-07-12 09:15:15.000000000","updated":"2021-07-19 13:05:35.000000000","submitted":"2021-07-19 13:04:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"800441-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"6eee1433be4db9d74877f2d494ad93729dc6788b","_number":800441,"virtual_id_number":800441,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-19 13:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"all":[{"value":1,"date":"2021-07-13 12:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2021-07-19 11:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-07-15 14:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2021-07-19 11:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 10:08:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-12 10:32:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 10:38:37.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-07-12 10:38:37.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-12 10:38:37.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-07-12 10:38:37.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b48f2d4002d6ef2fcb1e3b4be7883a057cff4ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-12 09:15:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa9796a8f3eee5e04be0ede9221ed56f3f94fb3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-12 09:19:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a0365e1b6aa837b101f4723029cd324d77a8ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 10:32:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d28221cba6f4098a70aba209bba30c8 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29b5a13ea5c34a719a289168339621eb : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0e6c51538ed4a0eafd086911b7e0ef4 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82c75cae2c74106a1dccb1baaf22516 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5181f9c279a94ee284a092460af450ec : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f9cfa9151b345a6861cc7e14c01b122 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72d5509dad4e4eb1985abf08f6b598fc : SUCCESS in 5m 03s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d61aa188dfe1422e95f1d820f02814fb : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c315fa0197e4f14a8ef75d3d50f1911 : SUCCESS in 41m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6bf766969bf46af9d61f60782e093b1 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":2},{"id":"80e1d53145b26168308033f0e8713ff58d0e1554","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-07-13 12:12:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nCode looking all good.","accounts_in_message":[],"_revision_number":2},{"id":"88c8702b97d751c7b6e514bd06f4251eec0c0792","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-15 14:51:19.000000000","message":"Patch Set 2: Code-Review+2\n\nNice","accounts_in_message":[],"_revision_number":2},{"id":"3eff1206bf82e0bdc9020e31fe858b6128e726ad","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-19 11:40:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57d74fa52d9ca877da3393b5efbf06cc68146a82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 11:41:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"397525dc6044e3c84ffec05b2fc332a89001c51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:04:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c087e9055e1415fbe1df32deecafa9e : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11bf87907cb640aeabcfd42c0fe8c959 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80a629394fe40c88188e91448f6fc1f : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad58062cedc435ebbea9a9b9d28c911 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d19c19645bf40f8b34e8fbff400e8e1 : SUCCESS in 4m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82b992ecaaa44d5fa12310dfef0b6105 : SUCCESS in 1h 22m 24s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a09105cc2a654abb96b0031ff61eb26b : SUCCESS in 37m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/63493581098246e39ef1c3ebd72ab32f : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ff55d54ab0fa2861124c2e7245b1a74edc3d5d2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:04:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6eee1433be4db9d74877f2d494ad93729dc6788b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:05:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b491c21796a7486e83ada3a91eb00f2d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d852cc7c862d748fe3a5285913e6df2fe32fc07","revisions":{"688e95a9ad1a5a2b41fd6f03fee8ba544700d361":{"kind":"REWORK","_number":1,"created":"2021-07-12 09:15:15.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/41/800441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/800441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/1"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-06-28 10:24:43.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-09 11:00:31.000000000","tz":120},"subject":"Add utils for parsing pkt processing cfg options","message":"Add utils for parsing pkt processing cfg options\n\nSee-Also: https://review.opendev.org/785236\nChange-Id: I8b1302bbdb57523f98b07b3ca88778d3a3103050\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688e95a9ad1a5a2b41fd6f03fee8ba544700d361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/688e95a9ad1a5a2b41fd6f03fee8ba544700d361"}]},"branch":"refs/heads/master"},"8d852cc7c862d748fe3a5285913e6df2fe32fc07":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-12 09:19:55.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/41/800441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/41/800441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/41/800441/2"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-06-28 10:24:43.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 09:19:04.000000000","tz":120},"subject":"Add utils for parsing pkt processing cfg options","message":"Add utils for parsing pkt processing cfg options\n\nAdd utils and constants for new Neutron OVS agent configuration\noptions:\n\n- resource_provider_packet_processing_without_direction\n- resource_provider_packet_processing_with_direction\n- resource_provider_packet_processing_inventory_defaults\n\nSee-Also: https://review.opendev.org/785236\nChange-Id: I8b1302bbdb57523f98b07b3ca88778d3a3103050\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8d852cc7c862d748fe3a5285913e6df2fe32fc07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8d852cc7c862d748fe3a5285913e6df2fe32fc07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
