)]}'
{"id":"openstack%2Fneutron-lib~801084","triplet_id":"openstack%2Fneutron-lib~master~I018963cbca6ce4c4e0045be0c78f9cd2aa942d45","project":"openstack/neutron-lib","branch":"master","topic":"bug/1936667","hashtags":[],"change_id":"I018963cbca6ce4c4e0045be0c78f9cd2aa942d45","subject":"Replace deprecated import of ABCs from collections","status":"MERGED","created":"2021-07-16 15:47:24.000000000","updated":"2021-07-20 16:16:01.000000000","submitted":"2021-07-20 16:14:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"801084-bug/1936667","meta_rev_id":"386e2e3f2e3e55f8d280100dade6e1cdf2590fe9","_number":801084,"virtual_id_number":801084,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 16:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-20 15:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-07-20 15:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 11:46:24.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-07-20 15:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 16:01:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-07-16 16:40:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-16 17:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 11:46:24.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-20 15:01:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e0a105834c0fd062951c69421b94b8042deece46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:47:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3db41d4b5d7fda0e508051c0921e2afe16b5219","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-16 16:01:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"421d05d0d192ea9aa8df95f5e0435f33da42ff5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 16:22:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d01d6f18701aaf0a251de2ba36c5222162fe5493","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 16:24:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"781235f8c7930b744955d3110aafff05c96ed46f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-16 16:40:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf547a85bed6c87f99a0b4b04c6dfaebe9b581ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:38:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea34cb817c8649e89e3b49e0a8b88f19 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128ec60c65e04833b16c912d6c287827 : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c478141db165418bb2715e3a4ee8966c : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688d4139e7214926a11d792c0d6bf6e0 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecaeda94f4ee41338e1357e6b8514360 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59e8e3ae16954b40b257d32a0afa7f79 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1645295ff46f413db4cd820030635132 : SUCCESS in 5m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91cfc0b8f07244f2ab05d62f1c775a99 : SUCCESS in 1h 14m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a2cbd5d92a1b419eaf9d37e588c690a0 : SUCCESS in 46m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5dc1f04f250416da5d6903ecbb940a7 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5f9c13ba4bf380e2c6b92817732e0ff5ed5aa07c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-17 03:23:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfcd01f3936ba98bae0e4db26d0707b7f7190de7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-17 03:24:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f4cb68d6ab58101f01bab658ed931d367d4b38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 04:47:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115a4aae8eed4465af5b152b0935bf1f : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4676f275a8334d6c9353a8abd716a80a : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50de2bb083f7431e8f9c2a7455717ffb : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4978773eb41d45dc860d9473102b99ec : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72657a9637c49a387cafbfab993c99f : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40624d661f94e588ab029c26f87b0ae : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e51d4ac375a4f01955913b5971bb1ad : SUCCESS in 4m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a50b6de04e34129b41b759c5a9cdf76 : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73f814c95e6d464d873f796583eac1ae : SUCCESS in 42m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6580400c8774294b5a17a1749481007 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":3},{"id":"5406d1dbc0f8f98b973188198dbce9c98e2dc733","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-20 11:46:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0c3dcd524d3729aa0aeaee13fdc1149460b1e4e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-20 15:00:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51291204d510daea52b15e71f7b30a5a2b1dd8d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 15:00:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8dce1bd18381260a9f87ab507430815d0a660e1e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-20 15:01:52.000000000","message":"Patch Set 3: Code-Review+1\n\nIndeed I saw that one in https://review.opendev.org/c/openstack/neutron/+/801068 logs, it will be nice to drop that deprecation log","accounts_in_message":[],"_revision_number":3},{"id":"98a2bf23ad69ab3d6f2375e26eab3a41cf4c0535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:14:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53090a7c062140f397cc2deed619c170 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b521c387b134a60aac4cf3d3055d365 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e196fa2166f409e9b54901bc6b6b03d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1af559c164be4ef6b20f2896d28668e7 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7942bb3cdf6940c3b84a02b812c91734 : SUCCESS in 4m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2108f379ff534056a777adba9fc0224c : SUCCESS in 1h 11m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/630e0b196c324d799b4f2e3cabac0e44 : SUCCESS in 42m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f8ea99b76ad4d2e8ec0d87589bb53df : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":3},{"id":"aac22a241491936f463e07dfa8e45c9666c4f44e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:14:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"386e2e3f2e3e55f8d280100dade6e1cdf2590fe9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:16:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fca93ae0954aae8cd9dd41fa2ab576 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e4b53b1fd1f5a7a32e32fb875592b44f9ba50497","revisions":{"f97f03054b5d72e92d96070393f7a7a8ba379555":{"kind":"REWORK","_number":1,"created":"2021-07-16 15:47:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/801084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/1"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:45:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:45:34.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: I018963cbca6ce4c4e0045be0c78f9cd2aa942d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f97f03054b5d72e92d96070393f7a7a8ba379555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f97f03054b5d72e92d96070393f7a7a8ba379555"}]},"branch":"refs/heads/master"},"2a43bbaf22a9e66988808dae74abbfadc944898c":{"kind":"REWORK","_number":2,"created":"2021-07-16 16:22:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/801084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/2"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:45:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 16:22:11.000000000","tz":0},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: I018963cbca6ce4c4e0045be0c78f9cd2aa942d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2a43bbaf22a9e66988808dae74abbfadc944898c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2a43bbaf22a9e66988808dae74abbfadc944898c"}]},"branch":"refs/heads/master"},"e4b53b1fd1f5a7a32e32fb875592b44f9ba50497":{"kind":"REWORK","_number":3,"created":"2021-07-17 03:23:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/801084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/84/801084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/84/801084/3"}}},"commit":{"parents":[{"commit":"44b3fc3f1821022aa5662b87a749b6b0eddaff20","subject":"Add a TODO note for resource classes constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/44b3fc3f1821022aa5662b87a749b6b0eddaff20"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:45:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-17 03:22:34.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: I018963cbca6ce4c4e0045be0c78f9cd2aa942d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4b53b1fd1f5a7a32e32fb875592b44f9ba50497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e4b53b1fd1f5a7a32e32fb875592b44f9ba50497"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
