)]}'
{"id":"openstack%2Fneutron-lib~802029","triplet_id":"openstack%2Fneutron-lib~master~I46381b70f770ccd32943644341388b09f7e14556","project":"openstack/neutron-lib","branch":"master","topic":"multi-ext-gw","hashtags":[],"change_id":"I46381b70f770ccd32943644341388b09f7e14556","subject":"multi-ext-gw: api-def and api-ref","status":"MERGED","created":"2021-07-23 14:10:38.000000000","updated":"2021-08-10 09:40:14.000000000","submitted":"2021-08-10 09:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"802029-multi-ext-gw","meta_rev_id":"3f2974491aeff68e60096e638a59b5c4df9f1b39","_number":802029,"virtual_id_number":802029,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 09:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-08-10 08:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-08-05 07:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},{"value":2,"date":"2021-08-09 13:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-08-10 08:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 14:17:06.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-23 15:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-05 07:08:31.000000000","updated_by":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"reviewer":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"state":"REVIEWER"},{"updated":"2021-08-10 08:15:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e76ac09a70b53a7818ab055d1405e9de874e5ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-07-23 14:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c84db9232f2dff973276a9360d0c62fec8ed6ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:42:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b241148ff0174b678f2d2c5b61560a48 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7e817b67ae4e5cbe6055ec59bf3c61 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b28d50a14fe4bdeaa34010ffc4a62aa : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7e8a22382d4245a4a5ac1870f20b1a : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d836a57e987245e48a4356eb3833ae57 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4173f120688e4216ac8868924d1a3b49 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29ed7cf702fb4f1ab31d10702b5d62c0 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ec8d8fc55f43349a6fb6e677c77eff : SUCCESS in 4m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8af197f29baa4114988f021a084430b5 : POST_FAILURE in 1h 29m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c081fda10cb4205887c81c73d77b284 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33464bb5da5a4e9fa5380a97ca120830 : SUCCESS in 44m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71d14b8ee7504dd9bf12982073a738e6 : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5bc737cecf809d4a6c790c8e1a50a47ccd5b2f3b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-07-23 17:45:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0236750d93c4d353b3ac7b2428b65f76cf245ea4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-27 14:49:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c50a8f105fe41c2033f156f2d74c208b98903a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 16:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/05b8cf6422044a7a977ddadbd101d1e0 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8882c55dd4a4e1d85ed4f94085cd96f : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5429c7cb69e4d2295c88afed7de389e : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c51a302b3f3f49a7bf708361c2877627 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a27ddf040854fa080cb7777f67b03a3 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a87395c7def4d619174f1834105ffef : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b0cc5b9b33f46ecb2a76e557242eff4 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c3955156f734cef99e8e4c9866f39a9 : SUCCESS in 4m 48s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2ede1451f1e4ef3a60e03514a052802 : SUCCESS in 1h 36m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d4c017c975464ead9121cce965caa2 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/723965a66bcf4d89a11c5a9b177b5183 : SUCCESS in 41m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b05db5e562e405b8c566068f8e24845 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3289f547f8883177526625b42bb08e9c36a51a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-08-03 08:52:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c67a926dcc9f7f27d991e23d7600a7ebebf6353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:30:37.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aedc675672f9476bb9d6b1ac35e0c224 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c89b1e166f6497f9849efacadd54433 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a328fd6445634e909f5c8f315d9853be : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c3f772de7d842598fcbde4db6dd72fa : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7b70a5e8db4a469c443e81e9135c69 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b5255d99c344beca2ace657facc0d28 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8c1ad278bab4d9fb0c6b63d716a573f : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/955b705e1f394627ae05f72079160f22 : SUCCESS in 4m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be68dad511414af6a5a233acb9edbd7c : SUCCESS in 1h 30m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b46de5d2a7294bfaa79531c88a961c5c : SUCCESS in 4m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7dfab4b108b249ab9e16160825fb569f : SUCCESS in 44m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de0aa325b436420baec090f2ac60d07c : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2c4e34539a98011b74c709622951dc04b4b59481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-08-03 14:41:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c12438f86f0c0b95fed9c6e77eb59dabcc1c1e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:35:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eb711d5fe5ac4ab39e387578c4886291 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee895d2778064ad5a8b00fbd4a9ccf61 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2334d4b0d3264269b7494a965028ee5e : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47ca1dc185254e7b9e5713fd57f93856 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fc018e4404431fa3258feb86c1eff0 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ad3f1e5de23445cbfcb98c76a4163ea : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f510ab4afada4142ad60b84b5f6a7a54 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf86d27045d48d99a837d6daa432f07 : SUCCESS in 4m 58s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b5f552f2b8d4986beb54d48f7bd5dd8 : FAILURE in 22m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ac50ef2db8408892c47f61a7a91525 : SUCCESS in 5m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e1f41063b85847da9e319be238c01fb8 : SUCCESS in 45m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b928df6d85e449fad5c95dd402dd111 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":3},{"id":"786b17040f9cf7f2b933053a5980c61375e1d150","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-08-04 07:18:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9343c66f4fa9d46a6281515960e23d994158119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 08:38:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62c43a97e19c41dba59894e0b22d3e79 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fa96802ac44a2992d0434f236bdebe : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f28d7f76a42c4dd29662989ae92c1076 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a2818b9cf9d4785a94afb8f9d00a4ee : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe10900388042408d90d99d1b434106 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a90c03fdc0a44bda2c09ea854dccaa0 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a49094d683ad48b4b2f228d91bc789a7 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c9d9165f8ec4fa5a8d2e51f11659b01 : SUCCESS in 4m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ceaa9d561414f00a4ae02a3dc199b96 : SUCCESS in 1h 18m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68a13bd2ea1c4374bcca654de430b7a1 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12ba05e38d9147899c035dde0e1ea061 : SUCCESS in 36m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f728acb94d24f59bcff84736221e245 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":3},{"id":"302d3a37c687511d4c097025aac5061f5f339bfc","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-08-05 07:08:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55856b12cc41aa06b7b0b9c46068a8935852bae7","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-08-06 09:39:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60f5b759ec4e0cdc0cd33b35f2722a2751d524b7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-09 13:02:33.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"555d964bfd632502cb404954baf45d7e75c641c5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-10 08:15:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b1633cd99b04bc9f1dc987c2927391126ecefbc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:15:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b9b809314adf3c85d9ed1ee3bd464877115679d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:37:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c898f61f14e44e0dba760026df895d98 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85630ba59e94eaa9cc8e4256003305e : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35ef86acf83441818662e2fbb4a4638a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e482d9f9a2d4884a78cf42eafc0ac82 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c8d342509ea40d89dbedd0cc3ac3a49 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42cb581cc8c64c978c121eb86bd0f86f : SUCCESS in 4m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbef16d604374d2290229790931eec63 : SUCCESS in 1h 18m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3559850234c54d34a0688f011279cb14 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/69716310076e486d8cc59ba24cf61a9c : SUCCESS in 51m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2643113aba543f6aee5f67e56b548c4 : SUCCESS in 48m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fd6819991acbec07e341b2356ee50f3be18f4a7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:37:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f2974491aeff68e60096e638a59b5c4df9f1b39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:40:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/789c3f647580426f8e21794634d1b8ae : SUCCESS in 2m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4ecd16f21642a6b5bb7d4be9b8a98e : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96b88fb9f2c47d3af793c04565bc946 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75057767eb98c302524e566c79e251ab334af8a1","revisions":{"17737e27882c7f5c5dc50c0bef212be8ac00f1f2":{"kind":"REWORK","_number":1,"created":"2021-07-23 14:10:38.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/29/802029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/802029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/1"}}},"commit":{"parents":[{"commit":"7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7f9d488ccf7a4fc00ac513bdaf9cda0c8774fb98"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-01-14 14:12:13.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-07-23 14:09:17.000000000","tz":120},"subject":"multi-ext-gw: api-def and api-ref","message":"multi-ext-gw: api-def and api-ref\n\nNew router attribute:\n    external_gateways\n\nNew router actions:\n    PUT add_external_gateways\n    PUT update_external_gateways\n    PUT remove_external_gateways\n\nChange-Id: I46381b70f770ccd32943644341388b09f7e14556\nPartial-Bug: #1905295\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/779511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/17737e27882c7f5c5dc50c0bef212be8ac00f1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/17737e27882c7f5c5dc50c0bef212be8ac00f1f2"}]},"branch":"refs/heads/master"},"da81178027cab22b8f36bd12cc50266ec8895bc2":{"kind":"REWORK","_number":2,"created":"2021-08-03 08:52:23.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/29/802029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/802029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/2"}}},"commit":{"parents":[{"commit":"60046e4f2c26d96986837004ac0b5642096959d5","subject":"Merge \"Use placement version allowing re-parenting RP update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60046e4f2c26d96986837004ac0b5642096959d5"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-01-14 14:12:13.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-08-03 08:51:46.000000000","tz":120},"subject":"multi-ext-gw: api-def and api-ref","message":"multi-ext-gw: api-def and api-ref\n\nNew router attribute:\n    external_gateways\n\nNew router actions:\n    PUT add_external_gateways\n    PUT update_external_gateways\n    PUT remove_external_gateways\n\nChange-Id: I46381b70f770ccd32943644341388b09f7e14556\nPartial-Bug: #1905295\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/779511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da81178027cab22b8f36bd12cc50266ec8895bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/da81178027cab22b8f36bd12cc50266ec8895bc2"}]},"branch":"refs/heads/master"},"75057767eb98c302524e566c79e251ab334af8a1":{"kind":"REWORK","_number":3,"created":"2021-08-03 14:41:54.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/29/802029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/29/802029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/29/802029/3"}}},"commit":{"parents":[{"commit":"60046e4f2c26d96986837004ac0b5642096959d5","subject":"Merge \"Use placement version allowing re-parenting RP update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60046e4f2c26d96986837004ac0b5642096959d5"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-01-14 14:12:13.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-08-03 14:02:35.000000000","tz":120},"subject":"multi-ext-gw: api-def and api-ref","message":"multi-ext-gw: api-def and api-ref\n\nNew router attribute:\n    external_gateways\n\nNew router actions:\n    PUT add_external_gateways\n    PUT update_external_gateways\n    PUT remove_external_gateways\n\nChange-Id: I46381b70f770ccd32943644341388b09f7e14556\nPartial-Bug: #1905295\nRelated-Change (spec): https://review.opendev.org/c/openstack/neutron-specs/+/779511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/75057767eb98c302524e566c79e251ab334af8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/75057767eb98c302524e566c79e251ab334af8a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
