)]}'
{"id":"openstack%2Fneutron-lib~803050","triplet_id":"openstack%2Fneutron-lib~master~I4fa575f726fc0a7b7066f749e3b3f90ecc520087","project":"openstack/neutron-lib","branch":"master","topic":"bug/1940518","hashtags":[],"change_id":"I4fa575f726fc0a7b7066f749e3b3f90ecc520087","subject":"Add network_ip_availability pagging and sorting support","status":"MERGED","created":"2021-07-30 15:14:34.000000000","updated":"2021-08-24 12:41:41.000000000","submitted":"2021-08-24 12:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"803050-bug/1940518","meta_rev_id":"7b39cffaf94415de1887c0e8e914c8cdb9258a3c","_number":803050,"virtual_id_number":803050,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 12:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-08-24 09:35:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-24 11:17:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-24 11:17:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-30 16:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 09:06:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-24 09:35:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-24 11:17:12.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"2b0b4ddb3728ce3876e79ec87a25deb20c673a39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-30 15:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58bb1577b4a9cf30ecbf1204e567ed5ec8bb4b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-30 15:22:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2009a4525a6b38aed93b23e6bcff163845804ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 16:35:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b777c0c0bb524cf1900ea169383f3550 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deaac2c126014f70813dcfdb3b196b97 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33f152ab54904959af7960566ef54ad4 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eda633b495f42edbbb0691c3adf2878 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e8e491d415845d7863b8e691e0e30f1 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f8539761ddc4c1a958b9d80ac0d8d90 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a779508e58f84c969d25860e8c10d436 : SUCCESS in 4m 23s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b60c4ced91d14cb3ab6f121dc9c4edba : SUCCESS in 1h 11m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5379483bbb514541ae28d8f9b43f9928 : SUCCESS in 37m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2698af6073641cbb33d634218814542 : FAILURE in 48m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2afe08087baeead62060af64edd9c1687e0fedba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-19 10:39:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e176588caa942d6679f378d69a40554f2820408d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:47:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2b34ddc0744424bc761c189b6204e0 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec50a5bdde44ed5925cb782fa7e0425 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29681b6fe6504f09b555e4c391704afd : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5c39dd8587497fad4f17818be0e449 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb75510673a34b26a84f61ca9656298e : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcbc46d374e0412dad4dd3833c0d41a7 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ba74cf58b8437c86dc1cd60e06a085 : SUCCESS in 4m 15s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc18da1568eb4fb4af5fa19c29a48037 : SUCCESS in 1h 04m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fcacffb8d7848f7aee48c0871d8edb4 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f65fabb522984d07a46005d9cef5fd6f : SUCCESS in 43m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6aa7f0700a9345649e77954d66bde14d : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":3},{"id":"a07f3e73b2c0210394aa3aa0fe4abaa666f854c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-19 13:44:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3bfc4d4c2784bc0c869c99d69af8ed90a135b3a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:06:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceddd765213e43009139af510927a033 : SUCCESS in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6efabaa152834783aa943ae193290ebd : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26089c44585d402cb2e4f3d61b36fa68 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eaaaa64c79e4b409610eb18ba78f695 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db79e296d3ad413c9ecf3b5f3bc4b37a : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7dd58992c66470a8490031134dc72f4 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0dcd0e540c041bcad74e930d7a784ff : SUCCESS in 6m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26228197230448f6865486d7c9905916 : SUCCESS in 1h 19m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcfafb6fd6c64ba8a8ac7129d58209dc : SUCCESS in 4m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a42186f9e4914d1e898ee5a5b4b85f36 : SUCCESS in 43m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28b0bccb79294ded8681cba1056db4c2 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e4a031fdc129ed675d772518101f1a41ffb4c96f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-20 14:26:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"67731e7b3dad4d3def0049e15848a762d4457bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:10:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409d8e025afe4113bcc2329443f07ab5 : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83836525e57c4307b51a6bff629b7e63 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6082d2efae56460cbf6cb0307d6ac6fd : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0705ef1d3b4e88b43c6ac9af2d918f : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0726afa66184741a5e207f4a2d60a25 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbd5cc2115324207b0a4e9294114054d : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b7c9af3fd8a4a3ba68cb555b6d22a1f : SUCCESS in 6m 01s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c1196830ef247b1afbdb347167f5f81 : SUCCESS in 1h 39m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9af2d2a2a1cf4c2e91e51e632d264be3 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/907adb4c2dae4bae87742089ff47697f : SUCCESS in 44m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b4b751f90174748ae84ac68e613f170 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ccdf9613249889ffe9172e5126e7acdc62af9a1a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-23 09:06:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c46c6ace2d5d9e471b296286c77db69ddcd2cb29","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-23 13:13:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11b926e0a5ad3bab476aa7c23cf36d68bd018847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-23 13:13:59.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"54b8ffaaa9b1ccf7272c2c4314ddb7cfacfe25db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 14:38:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb3fedf31144d8cbf8759185c69bfff : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b5103d5db9f4e50aec8390e171dcdbb : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeb1fda260844739a2d3d9fabe9fca09 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27351702584d462e892db8b22a92a409 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be4dd9ee562c4ae083bbe24c962002c1 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b85ce6fa14044413b16acfec29577ece : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d3a80711201477ab66641112c78c767 : SUCCESS in 6m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71f9d7f1adfe4c6eac3f412ce5c00ee0 : SUCCESS in 1h 19m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09eb5df706da41cc934e437ab9be9b21 : SUCCESS in 5m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/685e1d07519643f2a4e5e8cc5a015e12 : SUCCESS in 42m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ae5da37a81a4c3a9b4d6a388e25508d : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":6},{"id":"402803399003cdf5169baa155766f9129f0c27ee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-24 09:35:27.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ac69e4794e58220423486daded08e1feaf95ca61","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-24 11:17:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0ee53a7b2bb635ccfc6a71b5b9bed957b7e0377b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:17:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"431fdc74a0e8d0926b35a34cf19afcd17a7cbb3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:39:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42143c3623904c8c8ba264336491a997 : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e208ddf21aa47f1b61e3872482e3f03 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4954e99a9f454b31a62f85ebfcd3752b : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd3c047c4e8a42d09c53e1dda24576dc : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746dd72da67f4c74a7b26cbc5d2fcc24 : SUCCESS in 5m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c277ea261ff2479fa453344f2f3f7519 : SUCCESS in 1h 17m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15584e87cab645c89afc2aef3334ba78 : SUCCESS in 5m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aaf236458f43465fa71ec52abde69971 : SUCCESS in 40m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53d7c514a32447fcb6bd0388d5ab0dd6 : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":6},{"id":"f614d893cd830780c138b922bdef7a9c0de282ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:39:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7b39cffaf94415de1887c0e8e914c8cdb9258a3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:41:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faaa11d508484c73a3ab3de5ac579d4c : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab7e72b52714c64b4cf4005d7e0a9d0 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4c9bfc15b7c8ae3bf7091223f97fe8c714da3311","revisions":{"8a644357a4a68cf9926cbe5c416f59b9aae1bc5a":{"kind":"REWORK","_number":1,"created":"2021-07-30 15:14:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/1"}}},"commit":{"parents":[{"commit":"2b46c1f74cbf076fca1a100a978a3a8a23d0338c","subject":"Adds l3-ndp-proxy extension api definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b46c1f74cbf076fca1a100a978a3a8a23d0338c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-30 15:14:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-30 15:14:01.000000000","tz":0},"subject":"[WIP] Add network_ip_availability pagging support","message":"[WIP] Add network_ip_availability pagging support\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8a644357a4a68cf9926cbe5c416f59b9aae1bc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8a644357a4a68cf9926cbe5c416f59b9aae1bc5a"}]},"branch":"refs/heads/master"},"123b0f31796d9088e01ed52c6bfc26911ade187d":{"kind":"REWORK","_number":2,"created":"2021-07-30 15:22:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/2"}}},"commit":{"parents":[{"commit":"2b46c1f74cbf076fca1a100a978a3a8a23d0338c","subject":"Adds l3-ndp-proxy extension api definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b46c1f74cbf076fca1a100a978a3a8a23d0338c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-30 15:14:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-30 15:22:21.000000000","tz":0},"subject":"[WIP] Add network_ip_availability pagging support","message":"[WIP] Add network_ip_availability pagging support\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/123b0f31796d9088e01ed52c6bfc26911ade187d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/123b0f31796d9088e01ed52c6bfc26911ade187d"}]},"branch":"refs/heads/master"},"c14f8c10798c90f7ac0ae30b791a4b582945b0d3":{"kind":"REWORK","_number":3,"created":"2021-08-19 10:39:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/3"}}},"commit":{"parents":[{"commit":"2b46c1f74cbf076fca1a100a978a3a8a23d0338c","subject":"Adds l3-ndp-proxy extension api definition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2b46c1f74cbf076fca1a100a978a3a8a23d0338c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-30 15:14:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-19 10:39:38.000000000","tz":0},"subject":"Add network_ip_availability pagging support","message":"Add network_ip_availability pagging support\n\nAdd \"network_ip_availability\" pagging support, defining \"network_id\" as\nprimary key. This primary key is unique per register retrieved and is\nneeded by ``neutron.api.v2.base.Controller`` to create the pagination\nhelper.\n\nPartial-Bug: #1940518\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c14f8c10798c90f7ac0ae30b791a4b582945b0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c14f8c10798c90f7ac0ae30b791a4b582945b0d3"}]},"branch":"refs/heads/master"},"64afefa898e5ed3c0e2cc808f1446fb896cd8d09":{"kind":"REWORK","_number":4,"created":"2021-08-19 13:44:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/4"}}},"commit":{"parents":[{"commit":"13363b4d0d2be1bcc78f65b11892e62091208a54","subject":"Merge \"Introduce new bgp_associations API definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/13363b4d0d2be1bcc78f65b11892e62091208a54"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-19 13:39:12.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-19 13:39:12.000000000","tz":0},"subject":"Add network_ip_availability pagging support","message":"Add network_ip_availability pagging support\n\nAdd \"network_ip_availability\" pagging support, defining \"network_id\" as\nprimary key. This primary key is unique per register retrieved and is\nneeded by ``neutron.api.v2.base.Controller`` to create the pagination\nhelper.\n\nThis patch modifies an existing API extension instead of creating a new\none. However:\n- This modification does not affect the Neutron behaviour without a code\n  change in the \"network_ip_availability\" extension enabling the pagging\n  support. The Neutron code change implies a bump in the neutron-lib\n  version.\n- Adding a new extension to modify the current one is overengineering.\n\nPartial-Bug: #1940518\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64afefa898e5ed3c0e2cc808f1446fb896cd8d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/64afefa898e5ed3c0e2cc808f1446fb896cd8d09"}]},"branch":"refs/heads/master"},"32081b72fd3647eaa7540b46b3df2de20a7a4cb3":{"kind":"REWORK","_number":5,"created":"2021-08-20 14:26:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/5"}}},"commit":{"parents":[{"commit":"13363b4d0d2be1bcc78f65b11892e62091208a54","subject":"Merge \"Introduce new bgp_associations API definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/13363b4d0d2be1bcc78f65b11892e62091208a54"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-19 13:39:12.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-20 14:26:00.000000000","tz":0},"subject":"Add network_ip_availability pagging and sorting support","message":"Add network_ip_availability pagging and sorting support\n\nAdd \"network_ip_availability\" pagging and sorting support, defining\n\"network_id\" as primary key. This primary key is unique per register\nretrieved and is needed by ``neutron.api.v2.base.Controller`` to\ncreate the pagination helper.\n\nThis patch modifies an existing API extension instead of creating a new\none. However:\n- This modification does not affect the Neutron behaviour without a code\n  change in the \"network_ip_availability\" extension enabling the pagging\n  support. The Neutron code change implies a bump in the neutron-lib\n  version.\n- Adding a new extension to modify the current one is overengineering.\n\nPartial-Bug: #1940518\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/32081b72fd3647eaa7540b46b3df2de20a7a4cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/32081b72fd3647eaa7540b46b3df2de20a7a4cb3"}]},"branch":"refs/heads/master"},"4c9bfc15b7c8ae3bf7091223f97fe8c714da3311":{"kind":"REWORK","_number":6,"created":"2021-08-23 13:13:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/803050/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/803050/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/803050/6"}}},"commit":{"parents":[{"commit":"13363b4d0d2be1bcc78f65b11892e62091208a54","subject":"Merge \"Introduce new bgp_associations API definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/13363b4d0d2be1bcc78f65b11892e62091208a54"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-19 13:39:12.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-08-23 13:13:55.000000000","tz":0},"subject":"Add network_ip_availability pagging and sorting support","message":"Add network_ip_availability pagging and sorting support\n\nAdd \"network_ip_availability\" pagging and sorting support, defining\n\"network_id\" as primary key. This primary key is unique per register\nretrieved and is needed by ``neutron.api.v2.base.Controller`` to\ncreate the pagination helper.\n\nThis patch modifies an existing API extension instead of creating a new\none. However:\n- This modification does not affect the Neutron behaviour without a code\n  change in the \"network_ip_availability\" extension enabling the pagging\n  support. The Neutron code change implies a bump in the neutron-lib\n  version.\n- Adding a new extension to modify the current one is overengineering.\n\nPartial-Bug: #1940518\n\nChange-Id: I4fa575f726fc0a7b7066f749e3b3f90ecc520087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4c9bfc15b7c8ae3bf7091223f97fe8c714da3311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4c9bfc15b7c8ae3bf7091223f97fe8c714da3311"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
