)]}'
{"id":"openstack%2Fneutron-lib~803914","triplet_id":"openstack%2Fneutron-lib~master~If1f7b28184c740c30388f564b704ba42c8967527","project":"openstack/neutron-lib","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","hashtags":[],"change_id":"If1f7b28184c740c30388f564b704ba42c8967527","subject":"Allow PUT on direction param in QoS minimum pps rule","status":"MERGED","created":"2021-08-09 12:27:40.000000000","updated":"2021-08-16 15:12:29.000000000","submitted":"2021-08-16 15:11:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803914-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"98ffa3a55b01f0405d4dccb66f2ce9c28711541c","_number":803914,"virtual_id_number":803914,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 15:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-08-11 10:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-11 07:28:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-08-16 13:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-08-10 07:32:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-09 12:28:24.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-09 12:28:24.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-09 12:28:24.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-08-09 12:28:24.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-09 12:28:24.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-09 13:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5f341d2939a729d897e11188ff8b2cb13f56ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-09 12:27:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f4207845e72c3c4d5ad9db0195bf48bb58987eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:43:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83010a3ad6124ea7bc6d318884a523ec : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a3684d874fe428aa87caaeab7204ba7 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6cab23cc477412690f8cffe4a045582 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4c409591064cc2b9047f997eec43c5 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bb893c29a14c2787d33cf6bac147ca : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3546d435cb94cbd9c4c8a325b8050e3 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81218ab0513542c7b78aee9afd39bf36 : SUCCESS in 5m 14s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3305b3ff69d4399911fe77584aeb92e : SUCCESS in 1h 14m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2d041251f3d14755ae2de18e5f6ab52f : SUCCESS in 37m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d78e0a7e3104dcbaff1884c824ea3f1 : FAILURE in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"42819412fd69f5b34e45809cc3eb0cf4883c8274","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-09 13:47:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80fc18d345cb348db348b23e127338adf67a2867","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-09 13:51:11.000000000","message":"Patch Set 1: Code-Review-1\n\nYou also need to modify the api ref:\n- https://review.opendev.org/c/openstack/neutron-lib/+/802486/1/api-ref/source/v2/qos.inc#1145\n- https://review.opendev.org/c/openstack/neutron-lib/+/802486/1/api-ref/source/v2/samples/qos/minimum_packet_rate_rule-update-request.json#3","accounts_in_message":[],"_revision_number":1},{"id":"d3b777e50961b70030ed897de7853bd4164f3505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-09 14:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9285d99024d2cdd9bbe2cfd46086b7d43db89704","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-09 14:29:51.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e You also need to modify the api ref:\n\u003e - https://review.opendev.org/c/openstack/neutron-lib/+/802486/1/api-ref/source/v2/qos.inc#1145\n\u003e - https://review.opendev.org/c/openstack/neutron-lib/+/802486/1/api-ref/source/v2/samples/qos/minimum_packet_rate_rule-update-request.json#3\n\nSorry, I totally forgot about it. Fixed.","accounts_in_message":[],"_revision_number":2},{"id":"56ef1dfd1648cc9c2d8eb5e66ab8f57a80589e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:43:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e90c0cc7098b4983930ff85434f437eb : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28c68d0e6784f1aaccfd1948a28dc5c : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49148b2b1a364280ac9de4f5b048f868 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64d899329eb84141bb37b17481f1e8d6 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df27d1973324e39bbe433476888ab21 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8a0c4fdc5d49d1b824c476b823c12b : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bebe404904d4f919ac9ee9ded815bb6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e562d959ee214c1b80cd96f797caec49 : SUCCESS in 6m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10257c3ae0674a768e70f9ac69baf564 : SUCCESS in 1h 14m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68b1b3f159bb41b889fe8735ef0666f1 : SUCCESS in 38m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fd953860e044c03a1a950cb00dbea8c : FAILURE in 50m 12s","accounts_in_message":[],"_revision_number":2},{"id":"61b0d1e7290fae9a384cfd2cb151542706b96807","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-10 06:18:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20ce044a6f655411f689dd95fff30f49b112477b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-10 07:04:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9564798e92e6141f2234f6b8516ff67dcf4f02e2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-10 07:32:06.000000000","message":"Patch Set 2: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"820b43394775b2ddf9128e79ebc413e8f2dda67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 07:40:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b4d0aa4e964a448a80c6a38f3dc3e23a : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98433b23e4d4a6196d8a75f6a6a85be : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def2be7358eb4611a408998f42e16286 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48b4357239f14c2ead8cdd0d4d4731a4 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af946730cab44fab8006d16c31de792 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4517be4f0a7747cd9d64e6535438b6b1 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/337ecc16a2cf4907b824a54b75f1d732 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3fe689cf830465abf517fa42286f0c6 : SUCCESS in 5m 35s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d40e0ae4db2945fcbf148ad2d330ee02 : SUCCESS in 1h 15m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f2f32ceaf311405aabe3a1cf79401d57 : SUCCESS in 41m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0449eaeae1f49c4baa6413b3f50b1c2 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c8ce64303a62074952259a4c3286e0c3fbb1274c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-11 07:28:35.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update","accounts_in_message":[],"_revision_number":2},{"id":"748d04d968fe08cb54371a33f453fa7c7c506d7a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-11 10:37:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f25d164287898da9446e409e071799fad0562281","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-16 13:49:33.000000000","message":"Patch Set 2: Workflow+1\n\nI think we can merge this","accounts_in_message":[],"_revision_number":2},{"id":"3ccde462d32dc16e36d084f5e64b129eb9351b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 13:50:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc01f7d895694ba84782168043a364ca6aa74859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:10:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a70d17ee4d9d432ab082a40e17cca6ba : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f7e1a4a0b34d3f87c29f33a6d723c0 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/052cf189e2ea40fba72739eda235b0d6 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee952394ae54408696f5bacb2b7cd190 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92aa0b84247847d3bf7d66d1c40aa357 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff5de1966b6347fd8e15248c5f822f1d : SUCCESS in 4m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11751ca142c440cf84eb29f2d25baaa1 : SUCCESS in 1h 16m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ce2a439282c64f91996e4a829c709b36 : SUCCESS in 43m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b99f7a5055de4d14987d04f650abb8b3 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"059758dd791fe4a3f2f911f2ee34f180dca3d1ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:11:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"98ffa3a55b01f0405d4dccb66f2ce9c28711541c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:12:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1d99a0ed0c814b5e995e1fff8a1900e4 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcdde776ade04b6e9c07f91658c6ede1 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb4775cf8362588775467f4f92e6418f379369db","revisions":{"58929a8ffcd60cd765321e278cd19eac980c9cfc":{"kind":"REWORK","_number":1,"created":"2021-08-09 12:27:40.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/14/803914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/14/803914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/1"}}},"commit":{"parents":[{"commit":"8c473f4d325e44b50629c21113a8bf1b5060c8f1","subject":"Merge \"Add API extension definition for QoS minimum pps rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c473f4d325e44b50629c21113a8bf1b5060c8f1"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-09 12:07:40.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-09 12:22:27.000000000","tz":120},"subject":"Allow PUT on direction param in QoS minimum pps rule","message":"Allow PUT on direction param in QoS minimum pps rule\n\nBased on discussion in the previous review [1] and IRC [2], it was\ndecided to allow direction parameter in QoS minimum packet rate\nrule to be updated with a PUT request. This behavior matches current\nbehavior of other QoS rules.\n\nIf the rule to be updated is associated with a bound port, the request\nis going to be rejected by the server.\n\nSince direction-oriented and direction-less QoS min pps rules are\nmutually exclusive, the request is going to be rejected if it would\nresult in mixing direction-oriented and direction-less rules in a single\npolicy.\n\n[1] https://review.opendev.org/802486\n[2] https://meetings.opendev.org/irclogs/%23openstack-neutron/latest.log.html#t2021-08-09T07:56:39\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: If1f7b28184c740c30388f564b704ba42c8967527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/58929a8ffcd60cd765321e278cd19eac980c9cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/58929a8ffcd60cd765321e278cd19eac980c9cfc"}]},"branch":"refs/heads/master"},"fb4775cf8362588775467f4f92e6418f379369db":{"kind":"REWORK","_number":2,"created":"2021-08-09 14:27:53.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/14/803914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/14/803914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/14/803914/2"}}},"commit":{"parents":[{"commit":"8c473f4d325e44b50629c21113a8bf1b5060c8f1","subject":"Merge \"Add API extension definition for QoS minimum pps rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8c473f4d325e44b50629c21113a8bf1b5060c8f1"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-09 12:07:40.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-09 14:27:35.000000000","tz":120},"subject":"Allow PUT on direction param in QoS minimum pps rule","message":"Allow PUT on direction param in QoS minimum pps rule\n\nBased on discussion in the previous review [1] and IRC [2], it was\ndecided to allow direction parameter in QoS minimum packet rate\nrule to be updated with a PUT request. This behavior matches current\nbehavior of other QoS rules.\n\nIf the rule to be updated is associated with a bound port, the request\nis going to be rejected by the server.\n\nSince direction-oriented and direction-less QoS min pps rules are\nmutually exclusive, the request is going to be rejected if it would\nresult in mixing direction-oriented and direction-less rules in a single\npolicy.\n\n[1] https://review.opendev.org/802486\n[2] https://meetings.opendev.org/irclogs/%23openstack-neutron/latest.log.html#t2021-08-09T07:56:39\n\nRelated-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: If1f7b28184c740c30388f564b704ba42c8967527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fb4775cf8362588775467f4f92e6418f379369db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fb4775cf8362588775467f4f92e6418f379369db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
